)]}'
{"id":"openstack%2Fdragonflow~675850","triplet_id":"openstack%2Fdragonflow~master~I8874ce4562dd9a2ba069b0a110a23f7b9f519fc2","project":"openstack/dragonflow","branch":"master","topic":"update_redis","hashtags":[],"change_id":"I8874ce4562dd9a2ba069b0a110a23f7b9f519fc2","subject":"Remove testing revision number in test_df_qos_driver","status":"ABANDONED","created":"2019-08-12 08:57:59.000000000","updated":"2019-08-12 20:57:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e6efd2d745d64a3c30b679bb99de32cbf0748c9e","_number":675850,"virtual_id_number":675850,"owner":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"value":-1,"date":"2019-08-12 17:40:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"value":2,"date":"2019-08-12 15:48:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2019-08-12 15:48:39.000000000","_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 10:38:01.000000000","updated_by":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"reviewer":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"state":"REVIEWER"},{"updated":"2019-08-12 10:38:10.000000000","updated_by":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"reviewer":{"_account_id":6598,"name":"Berezovsky Irena","email":"irenab.dev@gmail.com","username":"irenab"},"state":"REVIEWER"},{"updated":"2019-08-12 17:40:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe709da646b6c1adbe0323e64b467d8962adeadc","author":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"date":"2019-08-12 08:57:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74eaabe60710dbf61ce58e8b99a9166fbc60e85f","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-12 10:38:01.000000000","message":"Patch Set 1: Code-Review+2\n\nThis should have been done long ago ;)","accounts_in_message":[],"_revision_number":1},{"id":"652453a8d1d560c8fcf25d2914ae289a6da3ac20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 10:45:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/50/675850/1/check/openstack-tox-pep8/5abb2f3/ : SUCCESS in 4m 45s\n- openstack-tox-py27 https://logs.opendev.org/50/675850/1/check/openstack-tox-py27/ab55e9b/ : SUCCESS in 6m 08s\n- openstack-tox-py35 https://logs.opendev.org/50/675850/1/check/openstack-tox-py35/ebb1b25/ : SUCCESS in 5m 48s\n- openstack-tox-py36 https://logs.opendev.org/50/675850/1/check/openstack-tox-py36/4ebe448/ : SUCCESS in 5m 06s\n- openstack-tox-docs https://logs.opendev.org/50/675850/1/check/openstack-tox-docs/9c4816d/html/ : SUCCESS in 2m 45s\n- dragonflow-tox-lower-constraints https://logs.opendev.org/50/675850/1/check/dragonflow-tox-lower-constraints/8416602/ : SUCCESS in 8m 43s\n- dragonflow-dsvm-fullstack-redis https://logs.opendev.org/50/675850/1/check/dragonflow-dsvm-fullstack-redis/9d438de/ : FAILURE in 1h 39m 06s\n- dragonflow-dsvm-fullstack-etcd-zmq https://logs.opendev.org/50/675850/1/check/dragonflow-dsvm-fullstack-etcd-zmq/b8eb56e/ : FAILURE in 1h 40m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f7a92627b8dee35c010e12ae2fddeda222182421","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-12 13:48:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d60a8680e10cc598aff1f5a0636975b2e8c7ec22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:37:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/50/675850/2/check/openstack-tox-pep8/3578494/ : SUCCESS in 4m 18s\n- openstack-tox-py27 https://logs.opendev.org/50/675850/2/check/openstack-tox-py27/b2c5f17/ : SUCCESS in 6m 58s\n- openstack-tox-py35 https://logs.opendev.org/50/675850/2/check/openstack-tox-py35/aa26c19/ : SUCCESS in 6m 04s\n- openstack-tox-py36 https://logs.opendev.org/50/675850/2/check/openstack-tox-py36/12c8751/ : SUCCESS in 5m 35s\n- openstack-tox-docs https://logs.opendev.org/50/675850/2/check/openstack-tox-docs/7588150/html/ : SUCCESS in 5m 37s\n- dragonflow-tox-lower-constraints https://logs.opendev.org/50/675850/2/check/dragonflow-tox-lower-constraints/fdcca5b/ : SUCCESS in 10m 29s\n- dragonflow-dsvm-fullstack-redis https://logs.opendev.org/50/675850/2/check/dragonflow-dsvm-fullstack-redis/4fe73c9/ : FAILURE in 1h 36m 50s\n- dragonflow-dsvm-fullstack-etcd-zmq https://logs.opendev.org/50/675850/2/check/dragonflow-dsvm-fullstack-etcd-zmq/eeb60e5/ : FAILURE in 1h 41m 04s","accounts_in_message":[],"_revision_number":2},{"id":"78d77179e0ae243f77a398dd10e002e5d0f34ab2","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-12 15:48:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"64d69c996f17d07a8febf068f598929807585198","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 17:40:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b280a1cfce040dcafa8be4420eac4e2 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9385d73eac8847f3b37b247757eeda4b : SUCCESS in 5m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2912a2898d6f418cbba45b7d436211c5 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1131b5e8b23d40768874f9c2ef5fdc3b : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7465f5d5884f4f8393719c672b470abb : SUCCESS in 4m 14s\n- dragonflow-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b7a3866873748d8b5e98a10ba286404 : SUCCESS in 9m 16s\n- dragonflow-dsvm-fullstack-redis https://zuul.opendev.org/t/openstack/build/7d1dd15279604b67b4bbd18a63bdc6a1 : FAILURE in 1h 48m 51s\n- dragonflow-dsvm-fullstack-etcd-zmq https://zuul.opendev.org/t/openstack/build/19e374fdd3c74f50a019e234514b2afb : FAILURE in 1h 42m 02s","accounts_in_message":[],"_revision_number":3},{"id":"628abe41b12efcbec67ff1e144885425dda1c12a","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-08-12 20:57:04.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/638509/","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0e9e0cc92b6f44bbff72b8c95674da133bc3435d","revisions":{"ecde772851db4f322cbbe941c1388fbd4a4d8803":{"kind":"REWORK","_number":1,"created":"2019-08-12 08:57:59.000000000","uploader":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"ref":"refs/changes/50/675850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/50/675850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/1"}}},"commit":{"parents":[{"commit":"67bec7f5a2a8b20b46a67a72466f142cdb2c0894","subject":"Upgrade redis driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/67bec7f5a2a8b20b46a67a72466f142cdb2c0894"}]}],"author":{"name":"Omer Anson","email":"omer.anson@toganetworks.com","date":"2019-08-12 08:57:38.000000000","tz":180},"committer":{"name":"Omer Anson","email":"omer.anson@toganetworks.com","date":"2019-08-12 08:57:46.000000000","tz":180},"subject":"Remove testing revision number in test_df_qos_driver","message":"Remove testing revision number in test_df_qos_driver\n\nThis code tests Neutron behaviour of increasing the revision number.\nThis is not code that is controlled by Dragonflow. If (when) it fails,\nDragonflow cannot fix it generically.\n\nChange-Id: I8874ce4562dd9a2ba069b0a110a23f7b9f519fc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/ecde772851db4f322cbbe941c1388fbd4a4d8803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/ecde772851db4f322cbbe941c1388fbd4a4d8803"}]},"branch":"refs/heads/master"},"8b69e22a8f0e95a8a937b94455b96ab24bccacda":{"kind":"REWORK","_number":2,"created":"2019-08-12 13:48:32.000000000","uploader":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"ref":"refs/changes/50/675850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/50/675850/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/2"}}},"commit":{"parents":[{"commit":"67bec7f5a2a8b20b46a67a72466f142cdb2c0894","subject":"Upgrade redis driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/67bec7f5a2a8b20b46a67a72466f142cdb2c0894"}]}],"author":{"name":"Omer Anson","email":"omer.anson@toganetworks.com","date":"2019-08-12 08:57:38.000000000","tz":180},"committer":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-12 13:47:29.000000000","tz":180},"subject":"Remove testing revision number in test_df_qos_driver","message":"Remove testing revision number in test_df_qos_driver\n\nThis code tests Neutron behaviour of increasing the revision number.\nThis is not code that is controlled by Dragonflow. If (when) it fails,\nDragonflow cannot fix it generically.\n\nChange-Id: I8874ce4562dd9a2ba069b0a110a23f7b9f519fc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/8b69e22a8f0e95a8a937b94455b96ab24bccacda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/8b69e22a8f0e95a8a937b94455b96ab24bccacda"}]},"branch":"refs/heads/master"},"0e9e0cc92b6f44bbff72b8c95674da133bc3435d":{"kind":"REWORK","_number":3,"created":"2019-08-12 15:48:39.000000000","uploader":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"ref":"refs/changes/50/675850/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/dragonflow","ref":"refs/changes/50/675850/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/dragonflow refs/changes/50/675850/3"}}},"commit":{"parents":[{"commit":"67bec7f5a2a8b20b46a67a72466f142cdb2c0894","subject":"Upgrade redis driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/67bec7f5a2a8b20b46a67a72466f142cdb2c0894"}]}],"author":{"name":"Omer Anson","email":"omer.anson@toganetworks.com","date":"2019-08-12 08:57:38.000000000","tz":180},"committer":{"name":"Shachar Snapiri","email":"shachar.snapiri@huawei.com","date":"2019-08-12 15:48:25.000000000","tz":180},"subject":"Remove testing revision number in test_df_qos_driver","message":"Remove testing revision number in test_df_qos_driver\n\nThis code tests Neutron behaviour of increasing the revision number.\nThis is not code that is controlled by Dragonflow. If (when) it fails,\nDragonflow cannot fix it generically.\n\nChange-Id: I8874ce4562dd9a2ba069b0a110a23f7b9f519fc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/0e9e0cc92b6f44bbff72b8c95674da133bc3435d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/dragonflow/commit/0e9e0cc92b6f44bbff72b8c95674da133bc3435d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
