)]}'
{"id":"openstack%2Felection~1001560","triplet_id":"openstack%2Felection~master~Ic6d18a60b2113442ada0c19764d797153567c125","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2026-08-24 18:06:08.000000000","reason":"\u003cGERRIT_ACCOUNT_14482\u003e replied on the change","reason_account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-08-26 13:40:17.000000000","reason":"Change was submitted"}},"hashtags":["ac-apc-reviewers-changes","core-reviewers-as-ac-apc"],"change_id":"Ic6d18a60b2113442ada0c19764d797153567c125","subject":"Active core reviewers are also Active Contributors","status":"MERGED","created":"2026-08-19 19:01:02.000000000","updated":"2026-08-26 13:42:32.000000000","submitted":"2026-08-26 13:40:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001560","meta_rev_id":"202220bb3ba6200bd6006b3f7cb3ae1de59bd4db","_number":1001560,"virtual_id_number":1001560,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 13:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2026-08-24 18:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2026-08-26 13:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2026-08-26 13:32:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-19 19:09:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-24 17:37:52.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2026-08-26 13:32:39.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"7dba52581674fb8c629d37bafef85a5dba1705e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-19 19:01:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e655cc4222684bf6c87cc5cbc7b477da483f271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 19:09:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331cc4a80ea140959d108dc8580b65e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2827a8663d7b413486aca4193d55eb4b : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce39ac4e0a714f56899aad6253113c8b : SUCCESS in 1m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bfc954749714d4ab27d233dab2a6b51 : SUCCESS in 5m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9b8fd126f65742d29d771700e5a55f55 : SUCCESS in 3m 14s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/6868c472ff8d4819a7827fa8cb594de2 : SUCCESS in 3m 44s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/195d31e5d6454cdf96c3c9800f321b1e : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1e13473c2dd9e84b652aafcb29eae8f467ae5cc0","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-19 19:14:05.000000000","message":"Set Work In Progress\n\nI\u0027ve got a logic error in here which is substituting reviewers with change owners and then combining their entries as duplicates. Will fix shortly.","accounts_in_message":[],"_revision_number":1},{"id":"15fcc804b2f110b8c9afaac3a5c93a437436da05","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-19 20:39:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f828d9e8a50ff4880eef8f54a7f82797d1576e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 20:56:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf0e209c0f774f9db02f85601c6fc604\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be49a9e864b640d49c5bcf9a940956db : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a77290db0e7e4c0d8bb24e877a8536d3 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5d02475a3964e21a525a87fe528519a : SUCCESS in 3m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4b5102f5a2a7413681eff27794d8a273 : SUCCESS in 3m 21s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/9d325310d8b3480bb94da93f58b833fd : SUCCESS in 2m 23s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e4482bc417344966912a9d2211f7d603 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"a571a3cd8789ae11f92abfc264be9e3932451741","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-19 21:27:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ba76e746b2fffe00c258848d70f5097e3e66f862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 21:38:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c01034c911d4fa6ab69a48197841d4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/751945dbeaa14584ab3850829e7cc8ca : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2e1bed5a3954b8db1f4bebf647d1238 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29bdd75683a4407dbf30cdacc544415c : SUCCESS in 5m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/da9f55139cb84a109e28c92e3a1f2047 : SUCCESS in 3m 26s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/91b57b962e284fb89399b439af3d1be6 : SUCCESS in 3m 00s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/0cd86e8dc8574f8ea7d6ff9bd53dfa94 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":3},{"id":"b4daed9f9d8ec2f8116ce0a8f1c6c2f5d93cde70","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-19 21:48:20.000000000","message":"Set Ready For Review\n\nRunning the same canary exercise as on previous changes in the stack but appending `--no-core-reviewers` the output is unchanged, as intended.\n\nOmitting that new option the output now has expanded activity dates for some contributors, as well as including additional contributors who left core review votes on the last revision of the same changes but owned no changes themselves. Execution time increased by roughly 10% according to tox\u0027s summary (though this probably doesn\u0027t scale linearly with duration).\n\nOne thing to note is that since the electorate files themselves (E-mail list txt files) aren\u0027t explicitly sorted the orders of some entries are moving around but none are missing.","accounts_in_message":[],"_revision_number":3},{"id":"32542215cc3c1eaab1881bfc4b03b9ab6876a852","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2026-08-24 17:37:52.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"623fea42dd70abf2af7354f44a03ca9ccdddb77f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-24 17:43:38.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c99d88a1b8216e133610bb07230de793ee1022cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 17:57:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a65addbcd5b1465791d6696a9aaf113a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3f84c74fd3941c0bf0dd7451cee9fa3 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70511ef9ca7d4d0faa123462fafca216 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21ecd382cbfe4d5094b84d7a9dc5a1f4 : SUCCESS in 3m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4aea53f86edd49e3a0f5a2858f324991 : SUCCESS in 3m 29s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/b9a8086a42e94dd5a4be290a46ba57e9 : SUCCESS in 3m 50s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/cbc9b190fe0b44aa8d8d85edd0be3aa4 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":4},{"id":"0926d1474c610dec53a60e2810ba132b550274ef","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2026-08-24 18:06:08.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f6045788544b65a05927b0395c433c235a2e8583","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-25 17:16:13.000000000","message":"Hashtag added: ac-apc-reviewers-changes","accounts_in_message":[],"_revision_number":4},{"id":"4d56633a4984e44eeca6a7be37f288d2b68f326e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-25 17:16:31.000000000","message":"Hashtag added: core-reviewers-as-ac-apc","accounts_in_message":[],"_revision_number":4},{"id":"c795db98a42180c96e18019a4053941d8eaaf160","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-26 13:32:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f418a333a9d44a454cfa6e44bab655fe58c6b51d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 13:32:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a22a24c5679dbd44d9e931c5565f22de630af90f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 13:40:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/821b99b1700945ac876406481df9ef4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c6de4d3ff9748c68934d6b53c80be44 : SUCCESS in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0d2e1d02ece4b0a879fd7ad175c3675 : SUCCESS in 2m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9e6a01a40ce44cda087cc766b1e0ffe : SUCCESS in 3m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b53491d2a0614822b5dfa736e2ed041b : SUCCESS in 2m 06s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5be9b04fd3474b20bce3fdf27905505a : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":4},{"id":"80504f01f7623d1075ed4a199892116e98789d53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 13:40:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"202220bb3ba6200bd6006b3f7cb3ae1de59bd4db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 13:42:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea854a0e8c244a1ba51b9bb3dd829d75\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/bc57f551a83f453584826f2fdba40a1f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f961a64652c33acb85aea52ce035e81a8322e858","revisions":{"6adf9f9171621c124282ab3e7eacc2819c08f2cf":{"kind":"REWORK","_number":1,"created":"2026-08-19 19:01:02.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/60/1001560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/60/1001560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/60/1001560/1"}}},"commit":{"parents":[{"commit":"d4e1f48783330f802b39423a75ba6de7df743530","subject":"Add internal toggle to skip change incrementing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d4e1f48783330f802b39423a75ba6de7df743530"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-14 20:05:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-19 18:51:41.000000000","tz":0},"subject":"Active core reviewers are also Active Contributors","message":"Active core reviewers are also Active Contributors\n\nThis is the best-effort implementation of revisions made to the TC\nCharter in I01023c948b4db94019c8861def1dd9694705d4ab which\nrecognizes \"Individual Members who reviewed or approved [merged\nchanges] with CR+2, CR-2 or W+1\" though with the caveat that Gerrit\nonly makes the final values of votes on the last revision of a\nchange queryable, so CR-2 will never appear in merged changes.\n\nChange-Id: Ic6d18a60b2113442ada0c19764d797153567c125\nSigned-off-by: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6adf9f9171621c124282ab3e7eacc2819c08f2cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6adf9f9171621c124282ab3e7eacc2819c08f2cf"}]},"branch":"refs/heads/master"},"083fde0b7f00a0583c7d751c863186fcd3f7d26c":{"kind":"REWORK","_number":2,"created":"2026-08-19 20:39:15.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/60/1001560/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/60/1001560/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/60/1001560/2"}}},"commit":{"parents":[{"commit":"2104c4fd9a644d74f23207633c8143fae405f5e4","subject":"Pass email explicitly to record_contributor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2104c4fd9a644d74f23207633c8143fae405f5e4"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-14 20:05:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-19 20:37:47.000000000","tz":0},"subject":"Active core reviewers are also Active Contributors","message":"Active core reviewers are also Active Contributors\n\nThis is the best-effort implementation of revisions made to the TC\nCharter in I01023c948b4db94019c8861def1dd9694705d4ab which\nrecognizes \"Individual Members who reviewed or approved [merged\nchanges] with CR+2, CR-2 or W+1\" though with the caveat that Gerrit\nonly makes the final values of votes on the last revision of a\nchange queryable, so CR-2 will never appear in merged changes.\n\nChange-Id: Ic6d18a60b2113442ada0c19764d797153567c125\nSigned-off-by: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/083fde0b7f00a0583c7d751c863186fcd3f7d26c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/083fde0b7f00a0583c7d751c863186fcd3f7d26c"}]},"branch":"refs/heads/master"},"fe979262b76621a0ea6091cb39b8a51a7502c41f":{"kind":"REWORK","_number":3,"created":"2026-08-19 21:27:10.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/60/1001560/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/60/1001560/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/60/1001560/3"}}},"commit":{"parents":[{"commit":"19a962555481d593b0cdf0cb18e63950ef68a351","subject":"Pass details explicitly to record_contributor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/19a962555481d593b0cdf0cb18e63950ef68a351"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-14 20:05:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-19 21:26:29.000000000","tz":0},"subject":"Active core reviewers are also Active Contributors","message":"Active core reviewers are also Active Contributors\n\nThis is the best-effort implementation of revisions made to the TC\nCharter in I01023c948b4db94019c8861def1dd9694705d4ab which\nrecognizes \"Individual Members who reviewed or approved [merged\nchanges] with CR+2, CR-2 or W+1\" though with the caveat that Gerrit\nonly makes the final values of votes on the last revision of a\nchange queryable, so CR-2 will never appear in merged changes.\n\nChange-Id: Ic6d18a60b2113442ada0c19764d797153567c125\nSigned-off-by: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fe979262b76621a0ea6091cb39b8a51a7502c41f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fe979262b76621a0ea6091cb39b8a51a7502c41f"}]},"branch":"refs/heads/master"},"f961a64652c33acb85aea52ce035e81a8322e858":{"kind":"REWORK","_number":4,"created":"2026-08-24 17:43:38.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/60/1001560/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/60/1001560/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/60/1001560/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/60/1001560/4"}}},"commit":{"parents":[{"commit":"19a962555481d593b0cdf0cb18e63950ef68a351","subject":"Pass details explicitly to record_contributor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/19a962555481d593b0cdf0cb18e63950ef68a351"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-14 20:05:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-24 17:43:20.000000000","tz":0},"subject":"Active core reviewers are also Active Contributors","message":"Active core reviewers are also Active Contributors\n\nThis is the best-effort implementation of revisions made to the TC\nCharter in I01023c948b4db94019c8861def1dd9694705d4ab which\nrecognizes \"Individual Members who reviewed or approved [merged\nchanges] with CR+2, CR-2 or W+1\" though with the caveat that Gerrit\nonly makes the final values of votes on the last revision of a\nchange queryable, so CR-2 will never appear in merged changes.\n\nChange-Id: Ic6d18a60b2113442ada0c19764d797153567c125\nSigned-off-by: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f961a64652c33acb85aea52ce035e81a8322e858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f961a64652c33acb85aea52ce035e81a8322e858"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
