)]}'
{"id":"openstack%2Felection~1001565","triplet_id":"openstack%2Felection~master~I3f09934cae41ed73abd06a86cfa927a439346826","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-08-26 08:48:06.000000000","reason":"Change was submitted"}},"hashtags":["ac-apc-reviewers-changes","core-reviewers-as-ac-apc"],"change_id":"I3f09934cae41ed73abd06a86cfa927a439346826","subject":"Pass details explicitly to record_contributor","status":"MERGED","created":"2026-08-19 20:39:15.000000000","updated":"2026-08-26 08:49:49.000000000","submitted":"2026-08-26 08:48:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001565","meta_rev_id":"51c0ca82b406b133ec26058e57ec705e74d05d07","_number":1001565,"virtual_id_number":1001565,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 08:48:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2026-08-26 08:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-08-24 20:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2026-08-26 08:39:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-19 20:48:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-24 20:22:03.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2026-08-26 08:39:36.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"9b3c700278b4e228afee2cff2e38352171ab0cd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-19 20:39:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7367babc4b13d4169bf0203f9ae6df8b1b61c76","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-19 20:44:48.000000000","message":"Patch Set 1:\n\nRunning the same exercise as on the parent change, the output is still identical so this stack so far should not change the makeup of the electorate.","accounts_in_message":[],"_revision_number":1},{"id":"b47a706791ebebda9c3706e74cdca4363f27a225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 20:48:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94e82438e3aa4e7db450dee671ed49ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/766ddf083c83408682e3ec69c8d323a5 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/552f082f435c4102935d7ee4715f034b : SUCCESS in 2m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f03f22a10ad4f03877327a040129547 : SUCCESS in 3m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0ecffd7d559e44b9ba06cedbdfd7d935 : SUCCESS in 4m 23s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/786d8419e4a84602bcb229a489b97516 : SUCCESS in 3m 25s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/6ebb017274e24a7690a275547f33cea9 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"36d4a86d98353fb03eeb5dcaa92c030581082378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-19 21:27:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1e1a95036a7c287b09372e9b4d701f3a8ad0cede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 21:37:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf56f60c9a5a49138e5adc718a4ea73f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f9892f8dc6949f4b8559a95b9056bbc : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b78c5308d1f444b29d5e30e4e07652a8 : SUCCESS in 2m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ad8ff3021124d0b8c48a70463e90299 : SUCCESS in 3m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e5493919969c42f2a8b798c8b8315d42 : SUCCESS in 4m 49s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/210688e544e6467c811676e56676a880 : SUCCESS in 3m 26s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/6c94d77da9b241a4af6eea5b196a1c34 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"1a16e78f144bb94773b5a73c92db66f54d5fe6cb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-19 21:38:48.000000000","message":"Patch Set 2:\n\nOutput is still unchanged testing locally with the same canary exercise.","accounts_in_message":[],"_revision_number":2},{"id":"3341afd1678d7ee9bcf6b970f77b73f7ca5a24dd","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2026-08-24 20:22:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ee09662dc04accc005617ac6630613177c53b9dd","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-25 17:16:13.000000000","message":"Hashtag added: ac-apc-reviewers-changes","accounts_in_message":[],"_revision_number":2},{"id":"84fef9c40876ac1b81dbc90c2bc30ad5513f6695","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-25 17:16:31.000000000","message":"Hashtag added: core-reviewers-as-ac-apc","accounts_in_message":[],"_revision_number":2},{"id":"4d9ad810866407a7ad245c6b193127476a8e9eaa","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-26 08:39:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a64c342f8b4715c740f3fd5b4668c31e6ddd31c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:39:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a48649b278d8a79e1cad85264ef10d2ede7571d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:48:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3be1af32fa9c471fb7395fdc2098062d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11facf22748e450fb2a2a54cc2db7c41 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/496d190cd88a44cdb8b07b16ea887b7a : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25705888b92a4959a0d452adb7af7bf1 : SUCCESS in 3m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/986d321641d94157bddc87d92b156e73 : SUCCESS in 2m 18s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/27c5826140f14a4e955ea787152ca550 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"1c68808396118abb65843c0c227f4f98b969969d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:48:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"51c0ca82b406b133ec26058e57ec705e74d05d07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:49:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c51a47dfd84f4d26b8e1a193d338e29f\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/d74b269c7d674814a0c4261ddb610b91 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"19a962555481d593b0cdf0cb18e63950ef68a351","revisions":{"2104c4fd9a644d74f23207633c8143fae405f5e4":{"kind":"REWORK","_number":1,"created":"2026-08-19 20:39:15.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/65/1001565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/65/1001565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/65/1001565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/65/1001565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/65/1001565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/65/1001565/1"}}},"commit":{"parents":[{"commit":"4d0defbb2f36f7c2b7df8350e56f1eab6934f00f","subject":"Add internal toggle to skip change incrementing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4d0defbb2f36f7c2b7df8350e56f1eab6934f00f"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-19 20:20:23.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-19 20:23:58.000000000","tz":0},"subject":"Pass email explicitly to record_contributor","message":"Pass email explicitly to record_contributor\n\nThis will make things easier later because E-mail will come from a\nsecond source besides just the change owner: we\u0027ll also have\nreviewer addresses and can\u0027t infer them from the change context\nwithout making unnecessary additional Gerrit queries.\n\nChange-Id: I3f09934cae41ed73abd06a86cfa927a439346826\nSigned-off-by: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2104c4fd9a644d74f23207633c8143fae405f5e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2104c4fd9a644d74f23207633c8143fae405f5e4"}]},"branch":"refs/heads/master"},"19a962555481d593b0cdf0cb18e63950ef68a351":{"kind":"REWORK","_number":2,"created":"2026-08-19 21:27:10.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/65/1001565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/65/1001565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/65/1001565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/65/1001565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/65/1001565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/65/1001565/2"}}},"commit":{"parents":[{"commit":"79cc65404a5a9d841dd6d033a07cbfffeb9a1e14","subject":"Add internal toggle to skip change incrementing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/79cc65404a5a9d841dd6d033a07cbfffeb9a1e14"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-19 20:20:23.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2026-08-19 21:26:29.000000000","tz":0},"subject":"Pass details explicitly to record_contributor","message":"Pass details explicitly to record_contributor\n\nThis will make things easier later because email, name and username\nwill come from a second source besides just the change owner: we\u0027ll\nalso have reviewer addresses and can\u0027t infer them from the change\ncontext without making unnecessary additional Gerrit queries.\n\nChange-Id: I3f09934cae41ed73abd06a86cfa927a439346826\nSigned-off-by: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/19a962555481d593b0cdf0cb18e63950ef68a351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/19a962555481d593b0cdf0cb18e63950ef68a351"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
