)]}'
{"id":"openstack%2Felection~677816","triplet_id":"openstack%2Felection~master~I0310c0cdbc6492f375e8a1364873ee80edda8e2b","project":"openstack/election","branch":"master","topic":"2019-september","hashtags":[],"change_id":"I0310c0cdbc6492f375e8a1364873ee80edda8e2b","subject":"Officially call it the \"U\" election","status":"MERGED","created":"2019-08-21 20:33:26.000000000","updated":"2019-08-22 20:22:09.000000000","submitted":"2019-08-22 20:22:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677816-1566505329156-b361f005","meta_rev_id":"55dce578d4738f6fdc42aa249d322675f24d6b8b","_number":677816,"virtual_id_number":677816,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-22 20:22:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":2,"date":"2019-08-22 20:11:29.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2019-08-22 00:13:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":1,"date":"2019-08-22 20:11:29.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 00:13:41.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-22 20:11:29.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-08-22 20:22:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0dbe22843e042ec6d0b95116bba6db0346b85bb0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-08-21 20:33:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64988f683264edd93ca28e9755c2f7007589c9c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 20:40:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4706d173d1364c81803a4d4f3daa88ae : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1146094a269c499d92c26b46f8d08376 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e58ae972e24c44c2a54adfcda5b58190 : SUCCESS in 3m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/461f46e7b2df47f79508d1ee482631b1 : SUCCESS in 4m 40s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/dda5550e16cd4c109900d7faea19c2c9 : SUCCESS in 3m 39s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a66433b43e97402c9fd04f2052e0fbe9 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"fc7fd0f122c74802790961ac37e4ffd9b9a9e515","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-22 00:13:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4c51a3edeeba270d8ba93a8b71ff4fe829e358a7","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-08-22 20:11:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72af77b731f8e2276054012c519d7e0301d638f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:11:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5502c11c42b76076872e23382b44ee1d1d3935b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:22:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea34657be6948f7befe1578e117e587 : SUCCESS in 3m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/246c2e4dc418466d87d26485a511159e : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/415091317bc644bfbcc014a8bd8f3b97 : SUCCESS in 5m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d53056e9cbcc457ea6e3f2b8e7178518 : SUCCESS in 5m 00s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5385e7be03f542feab09ca11f5b0f3b3 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"83f174b20bbfcd56d300444810d11d1df5fe7739","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:22:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d65bdcac7d43c2d50865f84edc3391c56e27158e","revisions":{"d65bdcac7d43c2d50865f84edc3391c56e27158e":{"kind":"REWORK","_number":1,"created":"2019-08-21 20:33:26.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/16/677816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/16/677816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/16/677816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/16/677816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/16/677816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/16/677816/1"}}},"commit":{"parents":[{"commit":"79c5a6d0373b6e2d293a70362a902afe0716b186","subject":"Merge \"Add dates for the \u0027U\u0027 Election.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/79c5a6d0373b6e2d293a70362a902afe0716b186"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-08-21 20:26:11.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-08-21 20:32:48.000000000","tz":0},"subject":"Officially call it the \"U\" election","message":"Officially call it the \"U\" election\n\nBecause it will still take some time for the OSF to perform\nlegal/trademark reviews on the results of the U cycle naming poll\nand we want to be able to move ahead with accepting nominations for\nupcoming elections, switch to just calling this cycle by the\nrequired initial letter we already know. In official communications\nwe can capitalize it as \"U\" but for the purposes of configuring our\nrelease automation the previous names were lower-case so use \"u\"\nhere instead.\n\nChange-Id: I0310c0cdbc6492f375e8a1364873ee80edda8e2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d65bdcac7d43c2d50865f84edc3391c56e27158e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d65bdcac7d43c2d50865f84edc3391c56e27158e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
