)]}'
{"id":"openstack%2Felection~677817","triplet_id":"openstack%2Felection~master~Ia34ea417202084cd92ee4f30f6acfd6a045e8f2c","project":"openstack/election","branch":"master","topic":"2019-september","hashtags":[],"change_id":"Ia34ea417202084cd92ee4f30f6acfd6a045e8f2c","subject":"Create \"U\" cycle candidate platform directories","status":"MERGED","created":"2019-08-21 20:33:26.000000000","updated":"2019-08-22 20:22:09.000000000","submitted":"2019-08-22 20:22:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677817-1566505329953-ebd307a2","meta_rev_id":"02e618e08ccf3a24fe447d30102e579d54cc48f5","_number":677817,"virtual_id_number":677817,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-22 20:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":2,"date":"2019-08-22 20:12:18.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2019-08-22 00:13:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":1,"date":"2019-08-22 20:12:18.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 00:13:52.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-22 20:12:18.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-08-22 20:22:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d72a27b1927a159219ffcc9c667e93b498d15c0b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-08-21 20:33:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a79afc7328cc48744864e5444985630e8c878f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 20:43:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f465858c08c4ba49eb4438251990209 : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/39a245fddd164c4186f4f7f5c6cebc29 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16ddc71749dc45949c1955c187f61168 : SUCCESS in 4m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/153e92a51dc944dfa9dca978d2cba44f : SUCCESS in 4m 05s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/7362fbc505974aa2a7d83c34865f1295 : SUCCESS in 3m 46s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/71d68c7a96bd4596be48109f2187060f : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"23247c7073d613acc2afcf39c31d8ee6bac943e8","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-22 00:13:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f7ee4dbc6ac591dd974b929d83518c068d084130","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-08-22 20:12:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f687e65df315998446e437e6b459b3989a434b3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:12:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aefe60f4399b996c30057c8935a9d89a4a1f3316","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:22:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"627f8d03596b32b2ad9c8c65d86995f9fd8e0adb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 20:22:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97efb152c0c0424ca905477122afdc19 : SUCCESS in 6m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2346c13e3e944f5fa63c9ad83f118c4d : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2f5015895e44fa0a7b764ecb3aa9013 : SUCCESS in 3m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/04f3b299fdc546a5abc4c4060576bf5f : SUCCESS in 2m 57s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/4635f1c80bec42da9e716cd17eff4c6b : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"600b2f2840275c4b95a33f0fd0c3a612e521c869","revisions":{"600b2f2840275c4b95a33f0fd0c3a612e521c869":{"kind":"REWORK","_number":1,"created":"2019-08-21 20:33:26.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/17/677817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/17/677817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/17/677817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/17/677817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/17/677817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/17/677817/1"}}},"commit":{"parents":[{"commit":"d65bdcac7d43c2d50865f84edc3391c56e27158e","subject":"Officially call it the \"U\" election","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d65bdcac7d43c2d50865f84edc3391c56e27158e"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-08-21 20:30:17.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-08-21 20:33:00.000000000","tz":0},"subject":"Create \"U\" cycle candidate platform directories","message":"Create \"U\" cycle candidate platform directories\n\nChange-Id: Ia34ea417202084cd92ee4f30f6acfd6a045e8f2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/600b2f2840275c4b95a33f0fd0c3a612e521c869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/600b2f2840275c4b95a33f0fd0c3a612e521c869"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
