)]}'
{"id":"openstack%2Felection~679881","triplet_id":"openstack%2Felection~master~I4791d15df6cdb3c50bfa2dd18ead5bc1337e7c28","project":"openstack/election","branch":"master","hashtags":[],"change_id":"I4791d15df6cdb3c50bfa2dd18ead5bc1337e7c28","subject":"Add Goutham Pacha Ravi candidacy for Manila PTL","status":"MERGED","created":"2019-09-03 23:44:16.000000000","updated":"2019-09-04 00:23:56.000000000","submitted":"2019-09-04 00:22:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679881-1567556558363-b0a5492a","meta_rev_id":"f231202ec6fcf2cb11b1260eb8e5cd4200b759bc","_number":679881,"virtual_id_number":679881,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2019-09-04 00:22:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2019-09-04 00:10:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-04 00:07:46.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2019-09-04 00:10:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 00:07:46.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-09-04 00:10:29.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2019-09-04 00:22:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"725ed73f9378d284f56d736f7fe166bed5c80248","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-09-03 23:44:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff0a9c280ea6a980d50c6e9454cf1844ad31386c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 00:04:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f169bc0839b34025b89858f4da3c5054 : SUCCESS in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d934a022d42a4b3294a0bd01977744c1 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7e8e113295d4593a206e6fff2c6dcf0 : SUCCESS in 3m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/03e3b0a1b2fe4ee1b84d771179c2eaea : SUCCESS in 4m 19s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/72c4416e011b4a0fa75c409d646e2dad : SUCCESS in 4m 11s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/d12540b46ff746a383aedb19d2b718a0 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d13320df5ab8522f641e69932866507ee74c2981","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-04 00:07:46.000000000","message":"Patch Set 1: Code-Review+2\n\nI confirm you are an eligible candidate for the office of Manila PTL.","accounts_in_message":[],"_revision_number":1},{"id":"73d9192376ba5bebc2507048a2467e6805607505","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2019-09-04 00:10:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI confirm that the candidate is eligible for Manila PTL","accounts_in_message":[],"_revision_number":1},{"id":"5e1b168f0300f9a44b085482348b9a0c15dd9a54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 00:10:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ad79ec65a3305d9d6eae9795bc937aab4b2c5c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 00:22:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c880e427000dbc4e43b529586fd88c631c4ebc5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 00:22:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c86545eb0d34905ac165cad7d8e81ae : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/425930039f3a47fea1834ae0876e0543 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d39bd77c50074d0a8439fbd78974c522 : SUCCESS in 3m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65f1c909ecd04f4e893c1c56b7644ab5 : SUCCESS in 4m 05s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/43f5405c9fa34bcd9c22a5d525686698 : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f231202ec6fcf2cb11b1260eb8e5cd4200b759bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 00:23:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-static https://zuul.opendev.org/t/openstack/build/a06448a8d76447d6a14886bdb78c372b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87236bf1058ebf87bf1887ceae093ce84bf64b73","revisions":{"87236bf1058ebf87bf1887ceae093ce84bf64b73":{"kind":"REWORK","_number":1,"created":"2019-09-03 23:44:16.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/81/679881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/679881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/679881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/679881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/679881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/679881/1"}}},"commit":{"parents":[{"commit":"606117490adea8eb2de61bf51f25efe4b69771aa","subject":"Merge \"Add Eric Fried candidacy for Nova U PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/606117490adea8eb2de61bf51f25efe4b69771aa"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-09-03 23:37:17.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-09-03 23:44:14.000000000","tz":-420},"subject":"Add Goutham Pacha Ravi candidacy for Manila PTL","message":"Add Goutham Pacha Ravi candidacy for Manila PTL\n\nChange-Id: I4791d15df6cdb3c50bfa2dd18ead5bc1337e7c28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/87236bf1058ebf87bf1887ceae093ce84bf64b73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/87236bf1058ebf87bf1887ceae093ce84bf64b73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
