)]}'
{"id":"openstack%2Felection~687181","triplet_id":"openstack%2Felection~master~I1e96789d51dc53433b329cd13d4204301b2f07f3","project":"openstack/election","branch":"master","topic":"transparent-verification","hashtags":[],"change_id":"I1e96789d51dc53433b329cd13d4204301b2f07f3","subject":"No longer match on repository short names","status":"MERGED","created":"2019-10-07 22:48:06.000000000","updated":"2019-11-21 21:33:10.000000000","submitted":"2019-11-21 21:33:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687181-1574371990998-a496186d","meta_rev_id":"3860820d3353bc11fb8b687cbb55c1a23604083c","_number":687181,"virtual_id_number":687181,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 21:33:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-11-06 03:01:50.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-11-06 02:57:32.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2019-10-14 07:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-14 04:40:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-11-06 02:57:32.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 04:40:20.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-10-14 07:51:09.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-11-06 02:57:32.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-11-06 03:01:50.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-11-21 21:33:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"deab25589d262d516f68c4f29d9edf5f7409210f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-10-07 22:48:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93bc60bdfde0d6711677c04b4869fc023fd21289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 22:55:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae1c34df07f7476f8b3fd2763f66a519 : SUCCESS in 4m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8d934af1018944068528b1d737fb9f74 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3c39415ea9a498ebec07068c6fe6f1e : SUCCESS in 3m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/343de2c97c1a40c2ae9409ed85893ecf : SUCCESS in 3m 49s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/637f6a733b6e4314bb34d8ab496e56d7 : SUCCESS in 4m 11s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a5d8f645f0db49a6aa2b311511c0db87 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8e0e43339c1717d8166ee772ea7baff4195c3782","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-10-10 06:28:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"35aa40fadf283c010d48f13b0b089d07e3aa963e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-10-11 12:48:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f56056e6a3116ce46d13022969e2741205dc32e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 13:03:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782b6b6681714296836852df673c22a3 : SUCCESS in 3m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3470108502794155b8f9ec9d2e41bc83 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34667cff341f40148d2b4c72e0c3d21f : SUCCESS in 3m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2b9ce2ccf9394c7995557783ba81e9ba : SUCCESS in 3m 40s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/0b9f802be3404142830f3ff144199bbe : SUCCESS in 3m 25s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/deac8bf1a551416b8587ced9824ed829 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"697fafa42d765a9578f7d4e3e4f7183eda5a9979","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-10-11 13:42:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a641b44bc9c1ae02e240e0fec403b0a96cf20f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 13:51:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d8ede2433546bdba42204d5887a0b2 : SUCCESS in 3m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/95cfc8b5d2a642f8a1559d46a7cf11f4 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6835821b616941588ed2b29807fd488d : SUCCESS in 4m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7fe963255edc4155bbd6ae8ad22d998d : SUCCESS in 3m 09s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/74a41a8479494d3090cc1bf11db25b29 : SUCCESS in 3m 39s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c8e05fbe9ea64ca1b9ef2b29b5638050 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":3},{"id":"e3451c5366cc2f6c508e12591d58aa946bb2a3c9","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-10-14 04:40:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ae7a71cdcaab53dce346d974039b4028922b51c2","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-10-14 07:51:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c85e58573b0ae3142e298eb23c5eb964a99effbf","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-11-06 02:57:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLooks good to me!","accounts_in_message":[],"_revision_number":3},{"id":"a4bada6b2efe202afc425c79a8d9330cebc0f944","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-06 03:01:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8ad63e468e5cd3bcce298eb06f89add45212da50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 21:23:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fc8ad8f2142c826ecb5b0f2ba31cfb6b0f9535de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 21:33:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e774884adfb54e5a884731a1b290d090 : SUCCESS in 3m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a9241c3074b545d1a0ed12ee3ec72fcd : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb4718641326464db546ac6f6d5d48e2 : SUCCESS in 4m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3d2e83a01a6b43689df04d034aeb6889 : SUCCESS in 3m 01s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b6e60becaf3248f287a9ed38893fb1c3 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f3060733cbc3911d5d1edcc185b723fc8581d0ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 21:33:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8b8f98aa4592cfce31ba6201dde1092d33311404","revisions":{"35f07c7b9d85d6a5832787e779e5cbfe41b047ab":{"kind":"REWORK","_number":1,"created":"2019-10-07 22:48:06.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/81/687181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/687181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/687181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/687181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/687181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/687181/1"}}},"commit":{"parents":[{"commit":"664d92bcc52c480836b94c7c98b39af4f3beed9b","subject":"Track oldest and newest Gerrit change numbers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/664d92bcc52c480836b94c7c98b39af4f3beed9b"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-10-07 22:36:39.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-10-07 22:36:39.000000000","tz":0},"subject":"No longer match on repository short names","message":"No longer match on repository short names\n\nWith the expansion of OpenDev\u0027s hosting, it is no longer guaranteed\nthat the last component of a repository\u0027s name is globally unique in\nthe OpenDev Gerrit deployment. Continuing to assume otherwise risks\nincorrectly including the wrong projects in electoral roll\ngeneration.\n\nChange-Id: I1e96789d51dc53433b329cd13d4204301b2f07f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/35f07c7b9d85d6a5832787e779e5cbfe41b047ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/35f07c7b9d85d6a5832787e779e5cbfe41b047ab"}]},"branch":"refs/heads/master"},"3fe74b4f4b303f4c5250c65736b915c53af3bab9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-11 12:48:59.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/81/687181/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/687181/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/687181/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/687181/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/687181/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/687181/2"}}},"commit":{"parents":[{"commit":"1b2ce90d0c18c680024c63d3afa38b19e9a9093e","subject":"Track oldest and newest Gerrit change numbers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1b2ce90d0c18c680024c63d3afa38b19e9a9093e"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-10-07 22:36:39.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-10-10 20:28:40.000000000","tz":0},"subject":"No longer match on repository short names","message":"No longer match on repository short names\n\nWith the expansion of OpenDev\u0027s hosting, it is no longer guaranteed\nthat the last component of a repository\u0027s name is globally unique in\nthe OpenDev Gerrit deployment. Continuing to assume otherwise risks\nincorrectly including the wrong projects in electoral roll\ngeneration.\n\nChange-Id: I1e96789d51dc53433b329cd13d4204301b2f07f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3fe74b4f4b303f4c5250c65736b915c53af3bab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3fe74b4f4b303f4c5250c65736b915c53af3bab9"}]},"branch":"refs/heads/master"},"8b8f98aa4592cfce31ba6201dde1092d33311404":{"kind":"NO_CHANGE","_number":3,"created":"2019-10-11 13:42:54.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/81/687181/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/687181/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/687181/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/687181/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/687181/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/687181/3"}}},"commit":{"parents":[{"commit":"646cd0d7c920eb72612609242cdb35ead7eaebbc","subject":"Track oldest and newest Gerrit change numbers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/646cd0d7c920eb72612609242cdb35ead7eaebbc"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-10-07 22:36:39.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2019-10-11 13:39:02.000000000","tz":0},"subject":"No longer match on repository short names","message":"No longer match on repository short names\n\nWith the expansion of OpenDev\u0027s hosting, it is no longer guaranteed\nthat the last component of a repository\u0027s name is globally unique in\nthe OpenDev Gerrit deployment. Continuing to assume otherwise risks\nincorrectly including the wrong projects in electoral roll\ngeneration.\n\nChange-Id: I1e96789d51dc53433b329cd13d4204301b2f07f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8b8f98aa4592cfce31ba6201dde1092d33311404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8b8f98aa4592cfce31ba6201dde1092d33311404"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
