)]}'
{"id":"openstack%2Felection~704703","triplet_id":"openstack%2Felection~master~If666499149c80e2194c33dfd17d115f39f2280bf","project":"openstack/election","branch":"master","topic":"publish-afs","hashtags":[],"change_id":"If666499149c80e2194c33dfd17d115f39f2280bf","subject":"Add promote-governance-election job","status":"MERGED","created":"2020-01-28 22:23:56.000000000","updated":"2020-01-28 23:24:56.000000000","submitted":"2020-01-28 23:21:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704703-1580253674006-728c2603","meta_rev_id":"feefc26d6a32964235a58a168930b18ba712757d","_number":704703,"virtual_id_number":704703,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 23:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-28 22:53:24.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2020-01-28 22:26:53.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2020-01-28 22:48:29.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 22:48:29.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-28 23:21:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f12018867a144f575d395aee5d3a689351b4dbb5","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-28 22:23:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0ae8f567ee847ab45f73be270d0ac401ac7949a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 22:24:11.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/election on branch master.  The error was:\n\n  Job promote-governance-election not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - build-openstack-docs-pti\n        - openstack-python35-jobs\n        - openstack-python36-jobs\n      check:\n        jobs:\n          - openstack-tox-linters\n          - election-tox-ci-checks-review\n          - election-tox-ci-checks-election\n      gate:\n        jobs:\n          - openstack-tox-linters\n          - election-tox-ci-checks-election\n      promote:\n        jobs:\n          - promote-tox-docs-static\n          - promote-governance-election\n\n  in \"openstack/election/.zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"0f2305cc63da851cbaabfb146609be29fd4d8d77","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-28 22:24:19.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2369e33f5de2bb8b229648f188ce2f0d97231059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 22:24:41.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change to a config project.\n\nThe syntax of the configuration in this change has been verified to be\ncorrect once the config project change upon which it depends is\nmerged, but it can not be used until that occurs.","accounts_in_message":[],"_revision_number":2},{"id":"6534a04054db3ea2ae8e03959ae8dcd593b45c90","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-28 22:26:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a96a8b9929b266e01b9debae599a5df729a644e4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-01-28 22:48:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"30de1d5ac2eb2144a53cb6f992c74911be7da691","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-28 22:53:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8f95582f52ce70294a61ea4fec3d1197f74eef3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 23:10:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3813ae63157440b4857de18cea5d1a6a : SUCCESS in 3m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8e254fe823384f16b8fc0d191a14777c : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a28588ffe0564ca882afa6b5f76543ef : SUCCESS in 5m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9f35028e7ed44f378ac549463ec13ffd : SUCCESS in 5m 34s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/93f6dd3e83ae48618088607bcf03bb09 : SUCCESS in 4m 08s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f14211a9aaa64ac5938c3b4ae9dae1b7 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"031c694ee9e3bb3ecddf36052fa47716c7b5f5d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 23:11:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3bc12c3e083ca44958b44b1bf7636aae4b376e17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 23:21:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1340188ad9be4e958b08ea7d89ffd47d : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ab713d890eb7480d922040fe063dd35b : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afd4d912666944f8922f4c28ed386197 : SUCCESS in 4m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97312e7e12854ff3be93800d181607e4 : SUCCESS in 5m 00s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e7ad8f59b6bc4ddfa94689f82b64b510 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"42c437b217d96d3587b30463403ac58f2434c500","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 23:21:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6213b693075c08769b625869d73365465372a3e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 23:24:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-static https://zuul.opendev.org/t/openstack/build/c309d038d8094c4bad662597fd02e2b2 : SUCCESS in 1m 37s\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/3fbbb905a6f0428d8a3e74d04d883feb : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"75b2cec7d3158c7e348d3c6d1f22f55f5bcde07f","revisions":{"db9f56c2e32dbcb6f6551dae5bbc7c9747559f8f":{"kind":"REWORK","_number":1,"created":"2020-01-28 22:23:56.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/03/704703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/03/704703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/03/704703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/03/704703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/03/704703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/03/704703/1"}}},"commit":{"parents":[{"commit":"d1d6c146e0d88cab619c016e0ae9f57afdbb095d","subject":"Fix get_series_data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d1d6c146e0d88cab619c016e0ae9f57afdbb095d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-28 22:19:14.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-28 22:23:52.000000000","tz":660},"subject":"Add promote-governance-election job","message":"Add promote-governance-election job\n\nThis job publishes the election site under the\ngovernance.openstack.org AFS volume.  This is our \"next generation\"\npublishing as it doesn\u0027t rely on storing things on a single static\nserver.  The current job, promote-tox-docs-static, will be removed\nafter we transition to serving the site from AFS.\n\nChange-Id: If666499149c80e2194c33dfd17d115f39f2280bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/db9f56c2e32dbcb6f6551dae5bbc7c9747559f8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/db9f56c2e32dbcb6f6551dae5bbc7c9747559f8f"}]},"branch":"refs/heads/master"},"75b2cec7d3158c7e348d3c6d1f22f55f5bcde07f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-28 22:24:19.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/03/704703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/03/704703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/03/704703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/03/704703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/03/704703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/03/704703/2"}}},"commit":{"parents":[{"commit":"d1d6c146e0d88cab619c016e0ae9f57afdbb095d","subject":"Fix get_series_data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d1d6c146e0d88cab619c016e0ae9f57afdbb095d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-28 22:19:14.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-28 22:23:59.000000000","tz":660},"subject":"Add promote-governance-election job","message":"Add promote-governance-election job\n\nThis job publishes the election site under the\ngovernance.openstack.org AFS volume.  This is our \"next generation\"\npublishing as it doesn\u0027t rely on storing things on a single static\nserver.  The current job, promote-tox-docs-static, will be removed\nafter we transition to serving the site from AFS.\n\nDepends-On: https://review.opendev.org/704702\nChange-Id: If666499149c80e2194c33dfd17d115f39f2280bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/75b2cec7d3158c7e348d3c6d1f22f55f5bcde07f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/75b2cec7d3158c7e348d3c6d1f22f55f5bcde07f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
