)]}'
{"id":"openstack%2Felection~715109","triplet_id":"openstack%2Felection~master~I9bd4af1577eef76f4524a0c47739f5b0c78de0f1","project":"openstack/election","branch":"master","topic":"puppet-openstack-v-ptl","hashtags":[],"change_id":"I9bd4af1577eef76f4524a0c47739f5b0c78de0f1","subject":"Add Shengping Zhong candidacy for Puppet OpenStack PTL","status":"MERGED","created":"2020-03-26 01:40:35.000000000","updated":"2020-03-27 18:39:24.000000000","submitted":"2020-03-27 18:36:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715109-1585334196832-f30010a2","meta_rev_id":"f2dcce64c7a693cd8340796b47056fae3ef88ab1","_number":715109,"virtual_id_number":715109,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"date":"2020-03-27 02:01:15.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-27 18:36:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2020-03-27 18:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2020-03-27 18:25:04.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2020-03-27 18:25:04.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 18:13:59.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-03-27 18:25:04.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-03-27 18:36:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12c4d671d500a787e760287be7d2dc82e551ba3c","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-26 01:40:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96201cf2af636856e9ff4d34e39fd84038539cd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 01:49:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e1f978f7efc4f218564170b15128de7 : FAILURE in 3m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/69b1872397b640cea6a0d3e33069893f : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3419aa7f29b543c593a66cfc616ba4a7 : SUCCESS in 3m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/96bdca504c504dd1b44c2887b5d41a2d : SUCCESS in 3m 26s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/a9c35ef286fa49f69f70688e97b58045 : SUCCESS in 2m 54s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/4c4e04da4af44c82a5eb7d171eb6f46c : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"888709f0ee77dafa4040b93fa3fbd9b382bbc851","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-27 00:55:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5bde58669a3dd65be445070fabdfaef5549ee8c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 01:03:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2f4a735d3642bfa5fa629be0ac2c3a : SUCCESS in 3m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/79a7c63743f34509a1fd80e8fe93b26f : FAILURE in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d490949bf1c4af58e861bf8bd129e4b : SUCCESS in 4m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/268accf270d44c9c9c8cd399827bd5ad : SUCCESS in 3m 56s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/e5f4b3b47d35437f969c27655ad88b3a : SUCCESS in 4m 03s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/85e4dd8efb734972bb70f233e54c4154 : FAILURE in 6m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2eca51d86f3a485c6f6ba6e961679785877e6a9a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-27 01:40:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c390dfc8ee3cf7c193a7524d840d8bd929e196cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 01:49:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a65411483f442eb456e0fc4ead2c87 : SUCCESS in 4m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ed7455787158455da1f1379aa94ce529 : FAILURE in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7098e3c012914f688e91995db2859e8a : SUCCESS in 3m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6a5e7cd3d8f848f09672a330b9b5b631 : SUCCESS in 3m 44s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/9a8a355c27cb489c889ec58412b81c5e : SUCCESS in 3m 42s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/d7e6d18ce69847799571c533d70cc800 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2a2a9f3e4f3946f886c9a30ef8a33bc126976fae","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-27 02:01:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8a9c9f3a9cdc89d61c9070bd25b3a03331a366ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 02:10:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8781053bfc984db0b02d611ff6597320 : SUCCESS in 3m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7839eee077054edf894b4d84195dfea1 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75494afbeb8343239ce520688369aa3f : SUCCESS in 3m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c7fefdc56b64ab88bd58e784e89a4dd : SUCCESS in 3m 47s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/22cc67f26e6c41268f96898f356af999 : SUCCESS in 4m 24s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/224f2da77bda41379a239e6b8b5a7497 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"37297654709569107f07d986fd8c2d94f893d07e","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-03-27 18:13:59.000000000","message":"Patch Set 1: Code-Review+2\n\nconfirm as a valid candidate","accounts_in_message":[],"_revision_number":1},{"id":"5dca679ae4d9f7470e08bf59ae918c233a1634a8","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-27 18:25:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9089c8643739097e3e1f7924abfef64c2798ebb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:25:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e03a83bb2a634e5a8606acdda2dbef0bd4b2ab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:36:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9163ab7a327a4f76b7dbad2e1471d267 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55ab5375be93437d9672c65eec34a922 : SUCCESS in 4m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5d19e1e8af64691a41c275ed599294c : SUCCESS in 3m 38s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f4459aeabb9b43638cd05e8855ef7abb : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d64ef4eaabfb9d4d0018858559a8ff76c6fa5e94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:36:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c3c0d8ec334ca9cb16544ee80b86808723ca6559","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:39:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/d275fd02d79c454e993823c25f5243b7 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35e9ed885627a07b1d9d6ffccbe342f6c374bfb9","revisions":{"35e9ed885627a07b1d9d6ffccbe342f6c374bfb9":{"kind":"REWORK","_number":1,"created":"2020-03-26 01:40:35.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/09/715109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/09/715109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/09/715109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/09/715109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/09/715109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/09/715109/1"}}},"commit":{"parents":[{"commit":"8259c8fc59b0f08665cf521a836da89743ab631f","subject":"Remove Jeremy Stanley from election officials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8259c8fc59b0f08665cf521a836da89743ab631f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-26 01:39:17.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-26 01:39:17.000000000","tz":480},"subject":"Add Shengping Zhong candidacy for Puppet OpenStack PTL","message":"Add Shengping Zhong candidacy for Puppet OpenStack PTL\n\nChange-Id: I9bd4af1577eef76f4524a0c47739f5b0c78de0f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/35e9ed885627a07b1d9d6ffccbe342f6c374bfb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/35e9ed885627a07b1d9d6ffccbe342f6c374bfb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
