)]}'
{"id":"openstack%2Felection~715204","triplet_id":"openstack%2Felection~master~I3c172a204470410221c496d9370cb2411de08293","project":"openstack/election","branch":"master","topic":"horizon-ptl-victoria","hashtags":[],"change_id":"I3c172a204470410221c496d9370cb2411de08293","subject":"Add Ivan Kolodyazhny candidacy for Horizon PTL","status":"MERGED","created":"2020-03-26 13:39:55.000000000","updated":"2020-03-27 18:40:57.000000000","submitted":"2020-03-27 18:37:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715204-1585334267879-795c03e7","meta_rev_id":"5959a5d450d1546ea5bc068354a835a1f23d41d1","_number":715204,"virtual_id_number":715204,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-27 18:37:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":2,"date":"2020-03-27 18:27:58.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":1,"date":"2020-03-27 18:27:58.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 18:27:58.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-03-27 18:37:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d84d3731006683475cc47871b674e5635c506be6","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-26 13:39:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16f9fb0773a6481bc552419ff790de9f015debf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 13:51:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb9d000fa104a4f91ae04a54459f68b : FAILURE in 4m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93655f8dc8a149ec884bcb70301f761d : FAILURE in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6744b9e824244d08223a798204e9fbd : SUCCESS in 5m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/66988486db444a2a9f7a608f0025d6df : SUCCESS in 3m 42s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/676b7dcd45ff4b81869a0457f7565cd3 : SUCCESS in 4m 18s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1bf765685d274d9cad7a45c9363e4aa8 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fbf98a406ced7533ebdec474c0c41eb87f73f0e1","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-26 19:08:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ac14daca90edfa7b2be0d196b29c2bbb6039e36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 19:20:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86686c3f9760465eaff832906deff724 : SUCCESS in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/de228212ce334fa1a123ca3e7f0cba1f : FAILURE in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/798eb174a16a4f06a603d114e8a6ceeb : SUCCESS in 4m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/22306828aa954291b4319c6b1f971c98 : SUCCESS in 3m 55s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/d5a82a0ac8734a9c94fc9edbd34e1123 : SUCCESS in 3m 37s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b663d7541caa49daa11a6c36a6ff41b6 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a8425d601b2f8a1da0cb6a32f1a17356e541db30","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-27 08:24:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee46ff1bcc9f88a0ed08d0033a038d13a605d9b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 08:32:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5bcb5ef8ea04e80abeeadd0ccd1e7ed : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0e6ce1dcd9594647bf6106afc1472d5a : FAILURE in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9032ebcfaf244d73ba6468985a71f3f6 : SUCCESS in 4m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5b1f4a18b8214364bff8b42fe3b0aad6 : SUCCESS in 4m 04s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/28d0c924822d4c5bb265c625b770d1b8 : SUCCESS in 5m 11s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2d52c192eed34af0a7b0db569de11294 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"67aa25ffd8c60762e397a7733f7459e7d7fbcffd","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-27 18:13:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"153109960689fa0272a42a6ace0a9c8ad6917ab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:27:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa6b8606586443ba832e03e6c92321e : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5ed236a082543949e0ba414e184eee1 : SUCCESS in 6m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/15fba0552919498eb12c423aeca52241 : SUCCESS in 4m 59s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/894a43c31da34aa49c47a48865fc7102 : SUCCESS in 4m 50s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c7d893fcf2ac42959e77e04a53e6b681 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"2f4bad4cff9c2d0f6d89e14ba84d21ef2a5aa36f","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-27 18:27:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nCandidacy is valid.","accounts_in_message":[],"_revision_number":2},{"id":"c2d830aea0bf9c1e1021cb793f489de4a710c2be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:28:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2014ea0c7c271089acd43a11dd49a817b4e0d943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:37:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/024588bfc5094c18a7f6596f9bffc39d : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de5b94444f834ea98c6bf0dfc2a535e4 : SUCCESS in 4m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/52e6ba87a84943b8a94be5ba138cc9d8 : SUCCESS in 5m 15s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/30402542f9dd47079d5538e16fa97cb4 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"72801b26bc422db0cf1316b487dcc454e3affb00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:37:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ad640db0cadff120fc62c25d20427a21f1d4bc0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:40:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/d8aa0c5cc6af4b01a0767ecf917e7754 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b1ae6bfac7bdd4ae76e34a004450bd1b9842c9a","revisions":{"84d7ac8f8b956334342d4d856e2d6079d1d2b8f7":{"kind":"REWORK","_number":1,"created":"2020-03-26 13:39:55.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/04/715204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/04/715204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/04/715204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/04/715204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/04/715204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/04/715204/1"}}},"commit":{"parents":[{"commit":"8259c8fc59b0f08665cf521a836da89743ab631f","subject":"Remove Jeremy Stanley from election officials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8259c8fc59b0f08665cf521a836da89743ab631f"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-26 13:39:42.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-26 13:39:42.000000000","tz":120},"subject":"Add Ivan Kolodyazhny candidacy for Horizon PTL","message":"Add Ivan Kolodyazhny candidacy for Horizon PTL\n\nChange-Id: I3c172a204470410221c496d9370cb2411de08293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/84d7ac8f8b956334342d4d856e2d6079d1d2b8f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/84d7ac8f8b956334342d4d856e2d6079d1d2b8f7"}]},"branch":"refs/heads/master"},"0b1ae6bfac7bdd4ae76e34a004450bd1b9842c9a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-27 18:13:20.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/04/715204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/04/715204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/04/715204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/04/715204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/04/715204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/04/715204/2"}}},"commit":{"parents":[{"commit":"82d4b0cc365495427a3113f31b80e8208da5e66a","subject":"Merge \"Add FengShengqin for Zun PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/82d4b0cc365495427a3113f31b80e8208da5e66a"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-26 13:39:42.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-27 18:13:20.000000000","tz":0},"subject":"Add Ivan Kolodyazhny candidacy for Horizon PTL","message":"Add Ivan Kolodyazhny candidacy for Horizon PTL\n\nChange-Id: I3c172a204470410221c496d9370cb2411de08293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0b1ae6bfac7bdd4ae76e34a004450bd1b9842c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0b1ae6bfac7bdd4ae76e34a004450bd1b9842c9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
