)]}'
{"id":"openstack%2Felection~715371","triplet_id":"openstack%2Felection~master~I9e8a5ba8a5189a3e1e8966c4310035f6e6242cb8","project":"openstack/election","branch":"master","topic":"mdulko-victoria-kuryr","hashtags":[],"change_id":"I9e8a5ba8a5189a3e1e8966c4310035f6e6242cb8","subject":"Add Michał Dulko candidacy for Kuryr in Victoria","status":"MERGED","created":"2020-03-27 09:34:34.000000000","updated":"2020-03-27 18:43:36.000000000","submitted":"2020-03-27 18:41:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715371-1585334509362-1fb5c97b","meta_rev_id":"4092df8eedd108a3f0bab297fbb13f44b86b6a0b","_number":715371,"virtual_id_number":715371,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-27 09:55:55.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"date":"2020-03-27 18:24:00.000000000","_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-27 18:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2020-03-27 18:19:57.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2020-03-27 18:19:57.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 18:19:57.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-03-27 18:24:00.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-03-27 18:41:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bcf5f02cb83fdf00f13d1f89f162a2d471874f8","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-27 09:34:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0d3cdff36609f96a78c2073cda20e1fc2770594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 09:45:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c0017b50aa54ee2b97e646c437a6c56 : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d1a6880c70684507a469b48bd465454e : FAILURE in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07701a09321d4a93b9540cb4b850260d : SUCCESS in 4m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e14c9343a3754c1cb1d6ec85bac27004 : SUCCESS in 4m 13s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2be8e65b03594b478bfd12884cbef591 : SUCCESS in 3m 41s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/8169646fb728493eae270b1e0da5d5cd : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"64da534861259d73bf0b4ae5af72efcc708f6c39","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-03-27 09:55:55.000000000","message":"Patch Set 1:\n\nrecheck\n\nSome deps issue?","accounts_in_message":[],"_revision_number":1},{"id":"7e4110a295d89d555ca355af7e675346e5d87b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 10:04:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7104bac15844a08bce15b14188ed30 : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ae32978e5a824123a7067f6cf1bf2d19 : FAILURE in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bafabc40a3a340cab5c00f81ec86bb49 : SUCCESS in 4m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b1f23a29814d426c9febde4ea05174d6 : SUCCESS in 3m 36s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2db85ac1de3b41d5a1a5e40f7b9a9ed4 : SUCCESS in 3m 53s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/69f3a24d9ee74a0fbce2234d21a2a678 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"83aa1eb4d9be748e2eac6ee6fd2936db5b2c09e5","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-03-27 18:19:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c075e834c41853d413263fc9870371d85ad25613","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-27 18:19:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nCandidacy is valid.","accounts_in_message":[],"_revision_number":1},{"id":"6391fd17c1623c79fe1773d243430d95188205e3","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-03-27 18:24:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"489212a925c2806e77c600418fc9fc2c7be11fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:32:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f12e644d0d409c8e3b54acc6ced1b8 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4163240330264e14b6d8670fbc4e56de : SUCCESS in 6m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8340b8420b334f5a893ae1e746f7c786 : SUCCESS in 3m 19s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/541ba551c6d343b6a7e31636de3dab0f : SUCCESS in 4m 43s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/68647a383f624e0bb487417a02a54fae : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5f8408f2be30f3c619f0acfc752b7759ed963797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:32:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"564138109606b465f04f6ae13a1d53a16e49bcc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:41:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e645b4e964494da89172179ea1e8f0 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c02173772dca4f6e887065df0c186267 : SUCCESS in 4m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f4564281ca2c41b5821a2da5ea5074e6 : SUCCESS in 3m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/458d2c3b58d14363b612c0a783f10dcc : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4b76d0cd87c8bef648a4a8ae461e703e4ad2b6a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:41:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"abc80e25f46e1c9fa84d7918f904b27e46626e55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:43:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/eb9322dca30f4153b4cf5273c0d7a984 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"326ed21e71968fdc2d1b6586647584cccd1fda4e","revisions":{"326ed21e71968fdc2d1b6586647584cccd1fda4e":{"kind":"REWORK","_number":1,"created":"2020-03-27 09:34:34.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/71/715371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/71/715371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/71/715371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/71/715371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/71/715371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/71/715371/1"}}},"commit":{"parents":[{"commit":"ebc55907692c5e948f1b4662aa7d36f01dde2e0c","subject":"Generate title lines based on length","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ebc55907692c5e948f1b4662aa7d36f01dde2e0c"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-03-27 09:33:53.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-03-27 09:33:53.000000000","tz":60},"subject":"Add Michał Dulko candidacy for Kuryr in Victoria","message":"Add Michał Dulko candidacy for Kuryr in Victoria\n\nChange-Id: I9e8a5ba8a5189a3e1e8966c4310035f6e6242cb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/326ed21e71968fdc2d1b6586647584cccd1fda4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/326ed21e71968fdc2d1b6586647584cccd1fda4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
