)]}'
{"id":"openstack%2Felection~715453","triplet_id":"openstack%2Felection~master~I204840bf4522b46daa12d4f564a1fff33eb6645d","project":"openstack/election","branch":"master","topic":"v-ptl-candidacy","hashtags":[],"change_id":"I204840bf4522b46daa12d4f564a1fff33eb6645d","subject":"Adding Slawek Kaplonski candidacy for Neutron","status":"MERGED","created":"2020-03-27 15:49:14.000000000","updated":"2020-03-27 18:36:37.000000000","submitted":"2020-03-27 18:31:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715453-1585333874711-f30010a2","meta_rev_id":"2d12502ff10ba5a5f6bfddae754de05e72ca7094","_number":715453,"virtual_id_number":715453,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-27 18:31:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2020-03-27 18:15:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2020-03-27 18:17:51.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2020-03-27 18:17:51.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 18:15:35.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-03-27 18:17:51.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-03-27 18:31:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92340de2a2125a550b6272db0955625c81e78e2f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-27 15:49:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b74d6e52a9c66b34ede3883169519e423352e8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 16:00:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70da38ab757b4b4787ab51fb8467ebdb : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d322d394fccd4ccb8c56bb3f95d00d4f : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c6d3b4fa2714f3fbf03041dbf4ef23d : SUCCESS in 6m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/24ce6d2ab8b445538d83ea8f0eede909 : SUCCESS in 6m 30s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/593759624e4b422890f92bcfaf87558f : SUCCESS in 5m 11s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/81223057038948d3840e79d119c0793c : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"83722744e87cf540f320333f97971df722b3b1a5","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-03-27 18:15:35.000000000","message":"Patch Set 1: Code-Review+2\n\nconfirm as a valid candidate","accounts_in_message":[],"_revision_number":1},{"id":"521a0f43623bf78571f73ecae9dfcd61f43efcb0","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-27 18:17:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nCandidacy is valid.","accounts_in_message":[],"_revision_number":1},{"id":"0f23d666d0b10c006ef35f018f26fbba4317b0a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:18:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c57619713bfba003eb52f85150fb7261b3b06a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:31:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec66db9c5ae50481ff20b564206448f7a97cb34e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:31:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd671c1a87c74441a4ae4b16e325a9a2 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01897dbc5c12435498f32c208ed839d2 : SUCCESS in 8m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/99cce6234d35497eb8a018cc803a3774 : SUCCESS in 5m 39s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/0811f1e4d5994b0abcd785c3490749b8 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d50339a8d8e11d70f0e592ec48bcdacfc329e113","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:36:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/d39a04f009bd4672b0c9f77979fb7bb2 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f1fe52f5468087634a05cb0060cee6dbbfe1e0d","revisions":{"9f1fe52f5468087634a05cb0060cee6dbbfe1e0d":{"kind":"REWORK","_number":1,"created":"2020-03-27 15:49:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/53/715453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/53/715453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/53/715453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/53/715453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/53/715453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/53/715453/1"}}},"commit":{"parents":[{"commit":"9cc1bcd83c02d367f31499e9e673bb0e808ee9be","subject":"Merge \"Update event table to work with simulateneous events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9cc1bcd83c02d367f31499e9e673bb0e808ee9be"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-27 15:48:43.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-27 15:48:43.000000000","tz":60},"subject":"Adding Slawek Kaplonski candidacy for Neutron","message":"Adding Slawek Kaplonski candidacy for Neutron\n\nChange-Id: I204840bf4522b46daa12d4f564a1fff33eb6645d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9f1fe52f5468087634a05cb0060cee6dbbfe1e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9f1fe52f5468087634a05cb0060cee6dbbfe1e0d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
