)]}'
{"id":"openstack%2Felection~715736","triplet_id":"openstack%2Felection~master~Ibc6e6f22cbf8c19c3114b6edd30969153cdc30eb","project":"openstack/election","branch":"master","topic":"victoria-qinling-ptl","hashtags":[],"change_id":"Ibc6e6f22cbf8c19c3114b6edd30969153cdc30eb","subject":"Add Lingxian Kong candidacy for Qinling PTL in Victoria","status":"MERGED","created":"2020-03-29 22:23:44.000000000","updated":"2020-03-30 16:57:25.000000000","submitted":"2020-03-30 16:53:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715736-1585587180649-68d14c44","meta_rev_id":"e00e567973d60d742e35208339416aaa1fc346a0","_number":715736,"virtual_id_number":715736,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-30 16:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2020-03-29 23:42:34.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2020-03-30 16:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-30 16:42:08.000000000","_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-29 23:42:34.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-03-30 16:42:08.000000000","updated_by":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"reviewer":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"state":"REVIEWER"},{"updated":"2020-03-30 16:43:41.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-03-30 16:53:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4085f3294ae30786ce8e855ea45da1cd58a6a2db","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-03-29 22:23:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df156c86572bdaa53f44b7a3736895cddf032690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 22:31:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d34ec2a7200f4b659035738ed2cf3539 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/854fc2ff0b4f4934b795dbef2ba25258 : SUCCESS in 4m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fd709a0d6070407fbd0485ba736cfabc : SUCCESS in 3m 43s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/bdb076f7bb6c4a3ea29143225c65c359 : SUCCESS in 3m 07s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f8365666baa941b8b19043577c695d2c : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"acd7fd2f3ffe2c2cb71e7a1c663ca5796d676bb2","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-03-29 23:42:34.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good to me.\n\nValid OSF member, and contributor to project.","accounts_in_message":[],"_revision_number":1},{"id":"1720d597ee75e94c3f498d62a0a5409dab184ba0","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2020-03-30 16:42:08.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"228a26259ed5fa9a9e53e9204cf07f7779cfd468","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 16:42:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ae63eb9b4f2827aea2298e0d512087d8a674f97","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-03-30 16:43:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b965fd995ee92aab310c1b9a73f98487fea5fb81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 16:53:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee1c752b61d4a0abea318980aa74d47 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a0909bfb87f4bb3a62038e4e6846654 : SUCCESS in 3m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e971026574f646128742b7d1f9affaf2 : SUCCESS in 4m 00s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ea7ad0b645af4cd0b66f63f3d89cab6e : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"19695f816b9a53e0c62aeba21346dd4951e135e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 16:53:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c27c85f4b09f05d7990fb30509d0a0217aa80c84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 16:57:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/55a737cd810f41e695b43d5609001950 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e048383161c55fd32713d4edbe6bff6a41cb345b","revisions":{"e048383161c55fd32713d4edbe6bff6a41cb345b":{"kind":"REWORK","_number":1,"created":"2020-03-29 22:23:44.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/36/715736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/36/715736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/36/715736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/36/715736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/36/715736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/36/715736/1"}}},"commit":{"parents":[{"commit":"7094885be1782cde1477f7cd2f1d3b03d6bc198d","subject":"Merge \"Add Michał Dulko candidacy for Kuryr in Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7094885be1782cde1477f7cd2f1d3b03d6bc198d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-29 22:23:34.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-03-29 22:23:34.000000000","tz":780},"subject":"Add Lingxian Kong candidacy for Qinling PTL in Victoria","message":"Add Lingxian Kong candidacy for Qinling PTL in Victoria\n\nChange-Id: Ibc6e6f22cbf8c19c3114b6edd30969153cdc30eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e048383161c55fd32713d4edbe6bff6a41cb345b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e048383161c55fd32713d4edbe6bff6a41cb345b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
