)]}'
{"id":"openstack%2Felection~716107","triplet_id":"openstack%2Felection~master~I59905e4c59068209607e8fa1360d9d9900f0bd46","project":"openstack/election","branch":"master","topic":"manila","hashtags":[],"change_id":"I59905e4c59068209607e8fa1360d9d9900f0bd46","subject":"Add Goutham Pacha Ravi candidacy for Manila PTL","status":"MERGED","created":"2020-03-30 23:18:37.000000000","updated":"2020-03-31 15:34:42.000000000","submitted":"2020-03-31 15:32:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716107-1585668754693-e19af6fb","meta_rev_id":"1d1bd5f8c8382f34a3f00a277eea1fc98c48db53","_number":716107,"virtual_id_number":716107,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 15:32:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2020-03-31 00:14:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2020-03-31 15:23:32.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2020-03-31 15:23:32.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 00:14:23.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-03-31 15:23:32.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-03-31 15:32:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33eeb4e3a0991a6dc7c26d02dbf493e20b6913bf","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-30 23:18:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc98713ba2eeb8bb61e963aab8c6f3a05a2cd9db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 23:27:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ed856d90744b1faffde7334169d320 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61272dd586364325846a790414df1d61 : SUCCESS in 2m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bfc546b7ac95431db6fe9ecbdca18ff2 : SUCCESS in 4m 33s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/396e6eb0325c4f0ba2f33137562d7d18 : SUCCESS in 2m 56s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5fbd3e97c5ab48389b86ae968f8098f5 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c234629656fdfb165749f788478c4d4d6dee7be7","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-03-31 00:14:23.000000000","message":"Patch Set 1: Code-Review+2\n\nconfirm as a valid candidate","accounts_in_message":[],"_revision_number":1},{"id":"a308f50b295247f304d58632620f91a67a8c7d6f","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-31 15:23:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nCandidacy is approved.","accounts_in_message":[],"_revision_number":1},{"id":"302633bc1990a22fae4d2c50ec8711a76cb1ba50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:23:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c983c6a88ac6754be3d6d6fa4c51527403314e7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:32:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b3d5deba506250fde9d0b18503c77295ef95f34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:32:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05a3998812c24f249d292cbaa5a675bc : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fb61b363b7445a9a4056a4c0fcb060a : SUCCESS in 5m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a72c72060cb9472295ed26ae55ae251c : SUCCESS in 3m 28s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/552ec6025cfe4ad0b58ac23de948a080 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9a14de465252da95768a445d25e027a0b87deee5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:34:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/49f8078ccd274fa88277817c21200a07 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b14fe1ba48c38e524e79709ac61ad7f2c98ebdf5","revisions":{"b14fe1ba48c38e524e79709ac61ad7f2c98ebdf5":{"kind":"REWORK","_number":1,"created":"2020-03-30 23:18:37.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/07/716107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/07/716107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/07/716107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/07/716107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/07/716107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/07/716107/1"}}},"commit":{"parents":[{"commit":"1af1c2beb8e58d027590ed348797ee9d40b631de","subject":"Merge \"Add Rong Zhu candidacy for Solum PTL in Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1af1c2beb8e58d027590ed348797ee9d40b631de"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-30 23:18:22.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-30 23:18:22.000000000","tz":-420},"subject":"Add Goutham Pacha Ravi candidacy for Manila PTL","message":"Add Goutham Pacha Ravi candidacy for Manila PTL\n\nChange-Id: I59905e4c59068209607e8fa1360d9d9900f0bd46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b14fe1ba48c38e524e79709ac61ad7f2c98ebdf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b14fe1ba48c38e524e79709ac61ad7f2c98ebdf5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
