)]}'
{"id":"openstack%2Felection~716323","triplet_id":"openstack%2Felection~master~Id65dd193ba26ac3452f28bc11a47241377462eb3","project":"openstack/election","branch":"master","hashtags":[],"change_id":"Id65dd193ba26ac3452f28bc11a47241377462eb3","subject":"Adding Balazs Gibizer candidacy for Nova PTL","status":"MERGED","created":"2020-03-31 14:51:11.000000000","updated":"2020-03-31 15:36:38.000000000","submitted":"2020-03-31 15:36:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716323-1585668998945-cfc8ac43","meta_rev_id":"dd18e4138dd0c59cee78525aafbc2ba05572e360","_number":716323,"virtual_id_number":716323,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"date":"2020-03-31 15:06:51.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 15:36:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":2,"date":"2020-03-31 15:24:42.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":1,"date":"2020-03-31 15:24:42.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 15:24:42.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-03-31 15:36:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"512928088d5e89aa6ded0b4866a80175a5e7902c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-03-31 14:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22079521645b2a71b232f2df9da04088ded0d9bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:04:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e6555dc325543a798f13080ac2557b8 : FAILURE in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e397338b018c4b32a516def7af10536c : SUCCESS in 6m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2169c23ce2ea4ccb9c2ffa8277df9d0a : SUCCESS in 6m 15s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/1142cb7645784434a19a37bc35c1e67d : FAILURE in 3m 27s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f312484875934fd4b1fc6403aa8a2ff7 : FAILURE in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ec74226351ed8d508c69946658d2813c380023ac","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-03-31 15:06:51.000000000","message":"Patch Set 1:\n\nrecheck fixed fundation membership","accounts_in_message":[],"_revision_number":1},{"id":"05c2e20640991610375a583a7dfb25e36be38ba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:17:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e26d14506f54db7b24e809dccbfac75 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca3986d561064827b41ea7520aac7d96 : SUCCESS in 5m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/518d2d87618e40c5848942dbf55ecf3b : SUCCESS in 4m 47s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/684b7073940345159e8d0e3357f74cf7 : SUCCESS in 4m 13s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a7aff838e4834d31bd2d256922a8adf1 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"89431d40ea3c1b52633df187c7b23a28a5b9baf9","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-31 15:24:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nCandidacy is approved.","accounts_in_message":[],"_revision_number":1},{"id":"b38e9724c42d463358a0d31069ac4287d3e0bb94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:24:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2bce3167591a8334337fa3e54b7ae46b86966ce1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:36:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59bfa4907b01472f9e1884f6c6a024d1 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d788f39f266f4643b184849e56987288 : SUCCESS in 6m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a0a41a22d4e741078dc2487dfddc4e16 : SUCCESS in 4m 03s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/61c48df716c34c5b96f9aec8885c040a : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"510b1d7b3371c56a9a22701e3eabead187c87b2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:36:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7759cd31d9b82246cf14bdc4b818e5d959822de8","revisions":{"7759cd31d9b82246cf14bdc4b818e5d959822de8":{"kind":"REWORK","_number":1,"created":"2020-03-31 14:51:11.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/23/716323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/23/716323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/23/716323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/23/716323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/23/716323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/23/716323/1"}}},"commit":{"parents":[{"commit":"3caa2400c69eed6d13f149de2cc1534c9b0eb093","subject":"Merge \"Adding Wes Hayutin candidacy for TripleO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3caa2400c69eed6d13f149de2cc1534c9b0eb093"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-03-31 14:48:32.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-03-31 14:50:50.000000000","tz":120},"subject":"Adding Balazs Gibizer candidacy for Nova PTL","message":"Adding Balazs Gibizer candidacy for Nova PTL\n\nChange-Id: Id65dd193ba26ac3452f28bc11a47241377462eb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7759cd31d9b82246cf14bdc4b818e5d959822de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7759cd31d9b82246cf14bdc4b818e5d959822de8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
