)]}'
{"id":"openstack%2Felection~716351","triplet_id":"openstack%2Felection~master~I72a2924ce0ecbf1a9ff31c5757ca55fb38fe65ab","project":"openstack/election","branch":"master","topic":"james-page-charms","hashtags":[],"change_id":"I72a2924ce0ecbf1a9ff31c5757ca55fb38fe65ab","subject":"charms: Add James Page for PTL","status":"MERGED","created":"2020-03-31 15:58:09.000000000","updated":"2020-03-31 20:14:57.000000000","submitted":"2020-03-31 20:12:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716351-1585685539220-73754ddb","meta_rev_id":"8b7eac9dabb30343597a0303426341eb8dcefc3d","_number":716351,"virtual_id_number":716351,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 20:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2020-03-31 20:02:13.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2020-03-31 20:02:13.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 16:16:41.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-03-31 20:02:13.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-03-31 20:12:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cbbdd88452a65016ee3c415fb754b5b47de6fe18","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2020-03-31 15:58:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e226782f147e23170b8b598f7ab3c9349008942","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 16:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb94730fe12b41b6a1499862dd1d0d06 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf0c5d54268e45bd8951660468be46e6 : SUCCESS in 5m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/69ca7bf6a2574bc798b2c3a5fcf63574 : SUCCESS in 5m 52s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/47e9b84607c541c18db5a17c20f9a232 : SUCCESS in 4m 04s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/6e02d712416b4244ae35a72c52d219e7 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"cea3b257756fa1412fa86559cad9e657dbc16fb8","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-03-31 16:16:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nConfirm as valid - +A for fast processing on the last day of nomination","accounts_in_message":[],"_revision_number":1},{"id":"7477685b76aa4f4c71c843665d4a74a0ff8f86ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 16:16:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2280e54234e6df3f034baf7cd4a39e14506d5a88","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2020-03-31 16:23:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe70c6f4ac10b751c9aa268c3287a74411c0c1b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 16:43:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3d824bbfe074cc29c9cc66f2438bccc : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7da77b22c17644e3af71ec93579cef55 : SUCCESS in 4m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/30f4270e65944349a29e9822e87f39ed : SUCCESS in 4m 50s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2ac5cfad0ef64985af665fd6f8c31c2c : SUCCESS in 5m 41s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1e315d965961486dac180349b3c8bb60 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"25e3427e342a57b28f4e2da43779bd3d9b11da35","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-31 20:02:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nCandidacy is valid.","accounts_in_message":[],"_revision_number":2},{"id":"85284d2d19b170333f3e2ff74ccd591bc2178045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 20:02:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"836500021ab513971b7739d98da5aa5e1840d881","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 20:12:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e04c6edee44113bdb4273230dd3bf9 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cbb252a51f34f6f80b7b1ec472b139d : SUCCESS in 4m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4c42eaf119854c149d5fcc0c1509a8dd : SUCCESS in 4m 25s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b02beddea72f4bee9ffe2ce0d04071c7 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1c5ca30e546870dcbe789716db1ea0a56c0dfafc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 20:12:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ac61a701c77e1b383d08fbd7b4a3bb8c92d41592","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 20:14:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/0c6494187ea746618bf9a089075a4815 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"935b6ebb80e0f468f04515c8a3a72cb03b898d48","revisions":{"c1398d9b47b36a2eced466aaabb8592c5d03dbd7":{"kind":"REWORK","_number":1,"created":"2020-03-31 15:58:09.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/51/716351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/51/716351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/51/716351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/51/716351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/51/716351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/51/716351/1"}}},"commit":{"parents":[{"commit":"147ee3e2410fa2964b5d7ab0ed1e2623e3af6e8e","subject":"Merge \"Add Masayuki Igawa candidacy for Quality_Assurance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/147ee3e2410fa2964b5d7ab0ed1e2623e3af6e8e"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2020-03-31 15:56:03.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@ubuntu.com","date":"2020-03-31 15:56:29.000000000","tz":60},"subject":"charms: Add James Page for PTL","message":"charms: Add James Page for PTL\n\nChange-Id: I72a2924ce0ecbf1a9ff31c5757ca55fb38fe65ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c1398d9b47b36a2eced466aaabb8592c5d03dbd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c1398d9b47b36a2eced466aaabb8592c5d03dbd7"}]},"branch":"refs/heads/master"},"935b6ebb80e0f468f04515c8a3a72cb03b898d48":{"kind":"REWORK","_number":2,"created":"2020-03-31 16:23:25.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/51/716351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/51/716351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/51/716351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/51/716351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/51/716351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/51/716351/2"}}},"commit":{"parents":[{"commit":"147ee3e2410fa2964b5d7ab0ed1e2623e3af6e8e","subject":"Merge \"Add Masayuki Igawa candidacy for Quality_Assurance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/147ee3e2410fa2964b5d7ab0ed1e2623e3af6e8e"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2020-03-31 15:56:03.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@ubuntu.com","date":"2020-03-31 16:21:39.000000000","tz":60},"subject":"charms: Add James Page for PTL","message":"charms: Add James Page for PTL\n\nChange-Id: I72a2924ce0ecbf1a9ff31c5757ca55fb38fe65ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/935b6ebb80e0f468f04515c8a3a72cb03b898d48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/935b6ebb80e0f468f04515c8a3a72cb03b898d48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
