)]}'
{"id":"openstack%2Felection~716525","triplet_id":"openstack%2Felection~master~If7041d7f18ce4f97b1d3993076cb6a4de920f2d2","project":"openstack/election","branch":"master","topic":"v_rally","hashtags":[],"change_id":"If7041d7f18ce4f97b1d3993076cb6a4de920f2d2","subject":"Add Andrey Kurilin candidacy for Rally PTL","status":"ABANDONED","created":"2020-04-01 08:19:37.000000000","updated":"2020-08-05 13:18:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"36c3ddaf0c0239f834e89eb5f3ceff849943b5ea","_number":716525,"virtual_id_number":716525,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-01 08:26:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":-2,"date":"2020-04-01 21:43:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-01 08:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-01 21:43:23.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"}],"messages":[{"id":"ffbd5ad86139cc90fa1fb078a3f9bcbb1d31ab27","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-04-01 08:19:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea0a0d45025ecf40a5e99766408e4d0a54c48466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 08:26:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d70f0cba0f9e402f8033281c59500821 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4f803f0dac44700a6eaabe36ca79688 : SUCCESS in 3m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c3975c8effa8435f99bf0ce3a73d19ce : SUCCESS in 3m 43s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/ba6913b22602432ea164232e4e238058 : SUCCESS in 3m 45s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/d026a3a3a54749fbafc254e3722ca03c : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7e5f91dbf1ac2290613ef78332ca0914d1fa7735","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-04-01 21:43:23.000000000","message":"Patch Set 1: Code-Review-2\n\nI think at this point, since you have missed the deadline and the TC will be appointing the PTL (likely you since you are interested :)) I will -2 this since we don\u0027t accept candidacies after the deadline.","accounts_in_message":[],"_revision_number":1},{"id":"dc3c5d0e5f224aa307f16742fc06c8ccb7d628a7","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-08-05 13:18:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87313b6819571ed0f4f330121258bcf372e2babd","revisions":{"87313b6819571ed0f4f330121258bcf372e2babd":{"kind":"REWORK","_number":1,"created":"2020-04-01 08:19:37.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/25/716525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/25/716525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/25/716525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/25/716525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/25/716525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/25/716525/1"}}},"commit":{"parents":[{"commit":"f58e3b8c9bd58b073d27d65f2ceb9af19002b627","subject":"Merge \"Add Feilong Wang candidancy for Magnum PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f58e3b8c9bd58b073d27d65f2ceb9af19002b627"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-04-01 08:19:31.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-04-01 08:19:31.000000000","tz":180},"subject":"Add Andrey Kurilin candidacy for Rally PTL","message":"Add Andrey Kurilin candidacy for Rally PTL\n\nChange-Id: If7041d7f18ce4f97b1d3993076cb6a4de920f2d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/87313b6819571ed0f4f330121258bcf372e2babd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/87313b6819571ed0f4f330121258bcf372e2babd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
