)]}'
{"id":"openstack%2Felection~717193","triplet_id":"openstack%2Felection~master~I6b9bb279952200df8e9aafb49756ebce6ecd1c8d","project":"openstack/election","branch":"master","topic":"victoria-election","hashtags":[],"change_id":"I6b9bb279952200df8e9aafb49756ebce6ecd1c8d","subject":"Close Victoria PTL election and post results","status":"MERGED","created":"2020-04-03 06:55:58.000000000","updated":"2020-04-15 00:12:22.000000000","submitted":"2020-04-15 00:10:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717193-1586909442949-b16e9fe7","meta_rev_id":"81e4c53bcf02bbeb26e8d3b9c0fce3533b68871b","_number":717193,"virtual_id_number":717193,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"date":"2020-04-08 02:19:42.000000000","_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 00:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"recommended":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2020-04-07 06:22:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":2,"date":"2020-04-15 00:00:32.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"date":"2020-04-08 02:04:03.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2020-04-15 00:00:32.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-03 07:06:02.000000000","_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 07:06:02.000000000","updated_by":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"reviewer":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"state":"REVIEWER"},{"updated":"2020-04-07 06:22:10.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2020-04-08 02:04:03.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-04-15 00:00:32.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-04-15 00:10:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e0376e43930ba695c429dbc5c01d1142175b7b2","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-04-03 06:55:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f861627a63ff70885daac13643a55429d5976833","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 07:07:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa77941d39e471e915e10bde0d0318a : SUCCESS in 5m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/690c4fde2df646ca9bf3543339df9d65 : SUCCESS in 4m 17s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/4fdd93ccc045439ebe4856ba370b5e23 : SUCCESS in 5m 50s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/0979471acd594809adf1356e8ddb01b5 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7a4bb5e519721826feefe597270caef25733c821","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-04-07 06:22:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cfbdf6b951d238397f381ed3d317287dcc884484","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-04-08 00:11:35.000000000","message":"Patch Set 1: Code-Review-2\n\nSo. I think we might have broken the close election script at some point.. the projects who didn\u0027t have candidates should still be listed in the file along with a tag saying that they need appointment by the TC. Looking into how to fix the script now.","accounts_in_message":[],"_revision_number":1},{"id":"a516944dcfbce16c7a6785746c318e7f9726601b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-04-08 01:08:16.000000000","message":"Patch Set 1:\n\n\u003e So. I think we might have broken the close election script at some\n \u003e point.. the projects who didn\u0027t have candidates should still be\n \u003e listed in the file along with a tag saying that they need\n \u003e appointment by the TC. Looking into how to fix the script now.\n\nYup we\u0027ve definitely broken the tools / the tools no longer match our current process.\n\nIn the past, in the event that a project(s) had no leader(s):\n\n * The officials would run close-election, and the db/yaml would have a bunch\n   of placeholders[1]\n * The officials would email the TC\n * The TC would \"fix\" the problem(s)\n * The officials would update the yaml/DB\n * The officials would use update-governance (which uses the yaml/DB)\n   to update the governance repo and at the same time publish the results\n   and call the election closed.\n\nNow we seem to:\n\n * The officials would run close-election, publish this change and call the\n   election closed\n * The officials would use update-governance (which uses the yaml/SB)\n   to update the governance repo with placeholders [2]\n\nNo doubt the new process is much easier for the officials, and probably the TC.  So we have a couple of options:\n\n * Take this change as is\n * Take this change as is and update the generated docs to have a list of \u0027Leaderless projects\u0027\n * Update the tools so that they more explicitly call out the leaderless projects.\n\nMy preference is the middle option\n\n\n[1] https://opendev.org/openstack/election/src/branch/master/openstack_election/cmds/close_election.py#L49-L54\n\n[2] https://opendev.org/openstack/election/src/branch/master/openstack_election/cmds/update_governance.py#L78-L82","accounts_in_message":[],"_revision_number":1},{"id":"d92ea5a81db250c78d32bb3c75dcdda05c6faf85","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-04-08 01:54:07.000000000","message":"Patch Set 1:\n\n@diablo_rojo \u0026 @tonyb: From my understanding https://review.opendev.org/#/c/717197/ patch is that actually election officials notify to TC via governance repo, and updating PTL election results to here like this is what election officials have been doing. For the election repo point of view, I think just updating election results would be sufficient. \n\nThrough this election, TC started to find leaderless projects earlier than we notified [1] so I think this process simplifies the steps. I also agree with mentioning leaderless projects via election repo by fixing scripts but seems that current process still works. Of course, we can enhance the scripts to explicitly describe the list of leaderless projects here.\n\n[1] https://etherpad.openstack.org/p/victoria-leaderless","accounts_in_message":[],"_revision_number":1},{"id":"e14544d7935437b0471cec69944dd5f163d72aef","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-04-08 01:56:55.000000000","message":"Patch Set 1:\n\nAnother simple proposal: adding here like this commit message would be nice?\n\n- Needed-by: I5ba004d6af2f9edb1707a6811fe25a2accc18467","accounts_in_message":[],"_revision_number":1},{"id":"8783dfb9d7aa62272d47b81437a24478d4f710c6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-04-08 01:57:55.000000000","message":"Patch Set 1:\n\n\u003e Another simple proposal: adding here like this commit message would\n \u003e be nice?\n \u003e \n \u003e - Needed-by: I5ba004d6af2f9edb1707a6811fe25a2accc18467\n\nI don\u0027t think that\u0027s needed","accounts_in_message":[],"_revision_number":1},{"id":"6c1a017f95a54f91959975569d550ef6f566cf9f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-04-08 02:04:03.000000000","message":"Patch Set 1:\n\n\u003e @diablo_rojo \u0026 @tonyb: From my understanding https://review.opendev.org/#/c/717197/\n \u003e patch is that actually election officials notify to TC via\n \u003e governance repo, and updating PTL election results to here like\n \u003e this is what election officials have been doing. For the election\n \u003e repo point of view, I think just updating election results would be\n \u003e sufficient.\n\nYou\u0027re correct *BUT* with my historical context this change \"looked wrong\" so I want to clarify the process (and what you have described matches what I described as the current process) so we\u0027re doing well there.\n\nThe follow-up would be to make the tools match the process, which they currently do not do.\n \n \u003e Through this election, TC started to find leaderless projects\n \u003e earlier than we notified [1] so I think this process simplifies the\n \u003e steps. I also agree with mentioning leaderless projects via\n \u003e election repo by fixing scripts but seems that current process\n \u003e still works.\n\nSo I interpret that as my \"Take this change as is and update the generated docs to have a list of \u0027Leaderless projects\u0027\" option?\n\n \u003e Of course, we can enhance the scripts to explicitly\n \u003e describe the list of leaderless projects here.\n \u003e \n \u003e [1] https://etherpad.openstack.org/p/victoria-leaderless","accounts_in_message":[],"_revision_number":1},{"id":"51c3351f2587fc6b691b0adb67ced84641bfef05","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-04-08 02:19:42.000000000","message":"Patch Set 1:\n\n@tonyb: Yes, and I additionally investigated - it seems that \"LEADERLESS\" projects were mentioned by Stein cycle [1] but did not explicitly mention from Train cycle [2][3].\n\n\u003e So I interpret that as my \"Take this change as is and update the generated docs to have a list of \u0027Leaderless projects\u0027\" option?\n\n[1] https://review.opendev.org/#/c/589676/2/doc/source/results/stein/ptl.yaml\n[2] https://review.opendev.org/#/c/644702/1/doc/source/results/train/ptl.yaml\n[3] https://review.opendev.org/#/c/679884/1/doc/source/results/ussuri/ptl.yaml","accounts_in_message":[],"_revision_number":1},{"id":"699e54ba34fee10a5b46ca04fe68a08f500fee9d","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-04-15 00:00:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b8ab32b92773215fd0d8eb9d5f55d29952b3fbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:00:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9bce0f3d57475800ba5976c7e29a9eba6d96cb0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:10:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d1680c94494fca9873ccff98d32801 : SUCCESS in 3m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/35ba51eb7666463a8515a7043f2f83bb : SUCCESS in 3m 33s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/d6d40001885347a8944cc4046ce581e0 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1f451ba465562d58aa98198688f89a9da492e28a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:10:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"35927a59f7dcfe0f0431b9b60ffce5a8d3f69cf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:12:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/c2a05b933dae4922b03d3cc626e9b4d3 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3191762ef865feee64635f8f3d47c8e60fcbefd3","revisions":{"3191762ef865feee64635f8f3d47c8e60fcbefd3":{"kind":"REWORK","_number":1,"created":"2020-04-03 06:55:58.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/93/717193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/93/717193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/93/717193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/93/717193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/93/717193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/93/717193/1"}}},"commit":{"parents":[{"commit":"f58e3b8c9bd58b073d27d65f2ceb9af19002b627","subject":"Merge \"Add Feilong Wang candidancy for Magnum PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f58e3b8c9bd58b073d27d65f2ceb9af19002b627"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2020-04-03 06:54:11.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2020-04-03 06:54:11.000000000","tz":540},"subject":"Close Victoria PTL election and post results","message":"Close Victoria PTL election and post results\n\nNomination period closed and voting is not needed so\nit is a good idea to post PTL election results.\n\nChange-Id: I6b9bb279952200df8e9aafb49756ebce6ecd1c8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3191762ef865feee64635f8f3d47c8e60fcbefd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3191762ef865feee64635f8f3d47c8e60fcbefd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
