)]}'
{"id":"openstack%2Felection~718317","triplet_id":"openstack%2Felection~master~Ie6a5ae8302dcd3ef9645067ba5d7f7bbd1c451f7","project":"openstack/election","branch":"master","topic":"victoria-ptl","hashtags":[],"change_id":"Ie6a5ae8302dcd3ef9645067ba5d7f7bbd1c451f7","subject":"Include Leaderless projects in the PTL series results","status":"MERGED","created":"2020-04-08 06:26:33.000000000","updated":"2020-04-15 00:14:18.000000000","submitted":"2020-04-15 00:10:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718317-1586909443998-c298d5ab","meta_rev_id":"18487d9ce4eaa567a70cef7e4af5b317738fd51c","_number":718317,"virtual_id_number":718317,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 00:10:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2020-04-08 10:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2020-04-14 23:59:04.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2020-04-14 23:59:04.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 10:50:44.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-04-14 23:59:04.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-04-15 00:10:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e24a89b0bbeba4dafa9926c30e599a7cf52399f7","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-04-08 06:26:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df91a20e8c9718208a36ac9d5bf22f36ec7e973c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 06:58:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd5a4060eec74996940debac5b1a3361 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dceb4a138c5449da1643d08acda38d9 : SUCCESS in 3m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b1b6d14d88da4de48b18b341d53f47a9 : SUCCESS in 4m 35s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/e570d1b9d99f4f0db90ded822473fcef : SUCCESS in 3m 46s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1cb867078975450d9ff5e615ab73f42f : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c7031389731a604b4e38054f587a404ed4c10f4c","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-04-08 10:50:44.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks great - thank u!","accounts_in_message":[],"_revision_number":1},{"id":"02f3a5595931d93505548658e071dfcf27009a48","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-04-14 23:59:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e189d56a1673fcb7c012e3ade5082cf4b4cb115","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:00:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b5e01aaccdf74e95d6bcdf7f4e225efbbc906d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:10:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d5f406fff0337bd63d005ff4ea345e56d557ef5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:10:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d2d063c7df4542b3c058aa8dd4d756 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4520c63d3ae149d1a2ff1c0c8bc448e4 : SUCCESS in 3m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8149210a7655447fb63d7dc52217750c : SUCCESS in 3m 38s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e04416401f704d34a82c4a7bf27e8e51 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1f559703ca220f4fbf79acf8cd1218b9321c3d27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:14:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/83647a76e3ad4f0ebd0b35896d98a309 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c297994e00179ba7377b64bc482a7c2bc7578a6","revisions":{"2c297994e00179ba7377b64bc482a7c2bc7578a6":{"kind":"REWORK","_number":1,"created":"2020-04-08 06:26:33.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/17/718317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/17/718317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/17/718317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/17/718317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/17/718317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/17/718317/1"}}},"commit":{"parents":[{"commit":"3191762ef865feee64635f8f3d47c8e60fcbefd3","subject":"Close Victoria PTL election and post results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3191762ef865feee64635f8f3d47c8e60fcbefd3"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2020-04-08 05:48:59.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2020-04-08 06:14:28.000000000","tz":600},"subject":"Include Leaderless projects in the PTL series results","message":"Include Leaderless projects in the PTL series results\n\nNow that we allow an election to close with one or more leaderless\nprojects it\u0027s worth tracking that on the results page.\n\nChange-Id: Ie6a5ae8302dcd3ef9645067ba5d7f7bbd1c451f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2c297994e00179ba7377b64bc482a7c2bc7578a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2c297994e00179ba7377b64bc482a7c2bc7578a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
