)]}'
{"id":"openstack%2Felection~734771","triplet_id":"openstack%2Felection~master~I3a040dd8986030d52eb3bd62b2ed862bcbdf6423","project":"openstack/election","branch":"master","topic":"drop_mock","hashtags":[],"change_id":"I3a040dd8986030d52eb3bd62b2ed862bcbdf6423","subject":"fix hacking error","status":"MERGED","created":"2020-06-10 09:07:42.000000000","updated":"2020-09-15 17:38:07.000000000","submitted":"2020-09-15 17:32:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734771-1600191166274-6df06640","meta_rev_id":"fc98d69b53a1f6472688260f7f9be002bfdb8541","_number":734771,"virtual_id_number":734771,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-15 17:32:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-06-10 09:37:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-15 17:24:11.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-15 17:24:11.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 09:37:19.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-15 17:24:11.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-09-15 17:32:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e298cd6e18402b7c28b36360c4c97ee623b4b924","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-10 09:07:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1fcc7b2dace89a7e2a42a34c028f92f97f1f77d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 09:23:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b89d459ee54acbb7bcf75feecdd45b : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8bdaffced2e4af5841022dc5a24a3ff : SUCCESS in 7m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f23cf42784ad4c2190bfaf4be701e958 : SUCCESS in 7m 15s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2c387b2572254294a59e88b1413f5a30 : SUCCESS in 7m 04s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5036e2df1dc843e286a160d2e52bc7c6 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3bf6f25a04854f718c5d5fc35e5fad953895800d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-10 09:37:19.000000000","message":"Patch Set 1: Code-Review+1\n\nI would blacklist E741 instead - but this works as well.","accounts_in_message":[],"_revision_number":1},{"id":"7af7074d1b109c4d8230920e70c6d8c1f77af12e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-15 17:24:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ec9601eaf4cd18ef3966d36e8d9d94fdfd55dab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 17:24:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e80aebfd56755e50ca60593200e624d8732a99d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 17:32:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"17425c6f9c6d7301ff9a04e19f042ecaaa91450b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 17:32:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd475f2fdf1f41998112dceecd084ed0 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a353fe7b8d5a4e0cb57c6028fbe01154 : SUCCESS in 3m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7688ee2ca6064532b06db1fb039ee674 : SUCCESS in 3m 36s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/bd1b4246595c4bdeaaa7f475e67fa832 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"18937161c62413fc8f88f1ec55659aba4264ac3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 17:38:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/58ed74f82f1943d6848db51755a38505 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e09d0ebbeb398b9efffa6269a25ea22cd094bd5f","revisions":{"e09d0ebbeb398b9efffa6269a25ea22cd094bd5f":{"kind":"REWORK","_number":1,"created":"2020-06-10 09:07:42.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/71/734771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/71/734771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/71/734771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/71/734771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/71/734771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/71/734771/1"}}},"commit":{"parents":[{"commit":"0a2ca581da735874bdd3b7dbce9d90465980173c","subject":"Merge \"Close the Victoria TC election\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0a2ca581da735874bdd3b7dbce9d90465980173c"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-10 09:07:07.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-10 09:07:28.000000000","tz":120},"subject":"fix hacking error","message":"fix hacking error\n\nChange-Id: I3a040dd8986030d52eb3bd62b2ed862bcbdf6423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e09d0ebbeb398b9efffa6269a25ea22cd094bd5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e09d0ebbeb398b9efffa6269a25ea22cd094bd5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
