)]}'
{"id":"openstack%2Felection~740181","triplet_id":"openstack%2Felection~master~I3ad364a4342904e7864af0723ea444d1ab770a06","project":"openstack/election","branch":"master","topic":"drop_future_imports","hashtags":[],"change_id":"I3ad364a4342904e7864af0723ea444d1ab770a06","subject":"update hacking version","status":"ABANDONED","created":"2020-07-09 08:11:51.000000000","updated":"2020-09-15 17:46:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d8dc8d35e374bbda9cecc74125f06ea9ec3108fd","_number":740181,"virtual_id_number":740181,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-09 08:24:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 08:24:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22bfb686ed6d27726364d1db0100406b7f7fbc03","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-09 08:11:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d8bc798266441239e349259fb19c657627647a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 08:24:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364a9697100a44caac7b0d61c927339b : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/251da0507f734974b57c55d67833f21f : SUCCESS in 4m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5d208788da184db8b6c91443ee4bf5f9 : SUCCESS in 5m 14s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/ff2e0946453d406f8317ed142250f7b1 : SUCCESS in 6m 19s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1bbb2cdd00254a4997437229c8c43c0b : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"766193df3f33ec34153103ddd2b9b877924face3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-15 17:46:08.000000000","message":"Abandoned\n\nThis is obsoleted by Ic0721e60f6583abb52bb665fbf113c8017318e2c (now merged).","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f4263e7f27f5573e785e29112586b5e0dcfea57","revisions":{"2f4263e7f27f5573e785e29112586b5e0dcfea57":{"kind":"REWORK","_number":1,"created":"2020-07-09 08:11:51.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/81/740181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/740181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/740181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/740181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/740181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/740181/1"}}},"commit":{"parents":[{"commit":"ad64f1a9d49924823fa380cbc509fabaa132ab9a","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ad64f1a9d49924823fa380cbc509fabaa132ab9a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-09 08:11:36.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-09 08:11:36.000000000","tz":120},"subject":"update hacking version","message":"update hacking version\n\nChange-Id: I3ad364a4342904e7864af0723ea444d1ab770a06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2f4263e7f27f5573e785e29112586b5e0dcfea57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2f4263e7f27f5573e785e29112586b5e0dcfea57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
