)]}'
{"id":"openstack%2Felection~753229","triplet_id":"openstack%2Felection~master~I35dd6952f2ee038a9d7937dca87f02ca3423fc01","project":"openstack/election","branch":"master","hashtags":[],"change_id":"I35dd6952f2ee038a9d7937dca87f02ca3423fc01","subject":"Add Artem Goncharov candidacy for SDK PTL","status":"MERGED","created":"2020-09-22 06:51:39.000000000","updated":"2020-09-23 19:40:40.000000000","submitted":"2020-09-23 19:39:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753229-1600889949612-dddb1e4e","meta_rev_id":"5b8f3fa217f19a4eea34f6d7e77b7c8763a898fb","_number":753229,"virtual_id_number":753229,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"date":"2020-09-22 12:13:31.000000000","_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-23 19:39:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2020-09-23 13:23:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-23 19:28:25.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-23 19:28:25.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 13:23:47.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-09-23 19:28:25.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-09-23 19:39:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c82889559301d079f15f916a8dd0cc24df22ea2f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-22 06:51:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b72ba464937cc79dcc6d8d22f77d09b81ca60303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 07:05:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6be8c8456284058952c4db8e307f02a : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e054af55ef043be8980232b205f335d : SUCCESS in 3m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/73ffe0b6519b456987c9849d4a94663c : SUCCESS in 5m 20s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/9185301797fc465dac28f29b6c47c8c5 : SUCCESS in 4m 22s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/35d1854a1cbe45a182189dcb8bf4ffe6 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b379959b5e9ace8c125861e5e503865fdc014718","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-22 12:04:11.000000000","message":"Patch Set 1:\n\nIt\u0027s okay to push this early, just be aware that we won\u0027t start confirming candidates until the nomination period begins in roughly 11.5 hours.","accounts_in_message":[],"_revision_number":1},{"id":"52f3a1928fcdef831fe61b64e26b7449db90e04c","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-22 12:13:31.000000000","message":"Patch Set 1:\n\nsure, missed the time part of the election start","accounts_in_message":[],"_revision_number":1},{"id":"68c51cbbb5df6ff36914767289c4194109ebb569","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-22 14:53:09.000000000","message":"Patch Set 1: Workflow-1\n\nPlease just be aware that election officials don\u0027t start confirming candidates until the nomination period begins soon (after ~0.5 days from now).","accounts_in_message":[],"_revision_number":1},{"id":"59c449e8cce056dabe6cf3eca35ce31faa25df46","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-23 06:20:08.000000000","message":"Patch Set 1: -Workflow\n\nrecheck\n\n(cannot access log - according to Notice, a failing log storage endpoint has been removed)","accounts_in_message":[],"_revision_number":1},{"id":"ad2c24dd6825318aeb210e2ddbafc31b90fb022c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 06:37:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3101cdf9c57347e2b8e15861c0a17ea4 : FAILURE in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f74258539d9f4474b085eb3e2af9cbe5 : SUCCESS in 4m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/22abec0e499744718b3a927178d766c7 : SUCCESS in 3m 53s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/e662b90e186a45c2a57e85aacef4fdf0 : FAILURE in 5m 20s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/41fbacef1e824388a91236e64edec134 : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d08a086cc26118464bd4a265552e1d876f2b067d","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-23 12:56:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f33755070ca417ac08ce90960256934b2f7fe0c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 13:19:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc49169d7684a1ab7c70b7cf7fa0d35 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e54e1a241c94d7983d7e61fb8a9a8d7 : SUCCESS in 5m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b35a775231ca4e57a8148e27961c60e5 : SUCCESS in 5m 18s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/aabad1d3bda94a36bdc76aef1bda0bed : SUCCESS in 5m 19s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b3c6914e9a5948b389af0f7e9bb41874 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5fcf135f68e5904c42cdc105555df6c0c31080f7","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-23 13:23:47.000000000","message":"Patch Set 1: Code-Review+2\n\nConfirms as a valid candidate","accounts_in_message":[],"_revision_number":1},{"id":"69238d41d0c406b3d1f0f796245521eb46d11456","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-23 19:28:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nhttps://openstack.org/community/members/profile/101132\nhttps://review.opendev.org/753409 merged on 2020-09-23 to openstack/openstacksdk for OpenStackSDK","accounts_in_message":[],"_revision_number":1},{"id":"335841ba07369efec44e2cbddc342340361d2036","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:28:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f9fc9407fc1446f12c9304fc9643521b09b8393","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:39:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4112be38dc78d4e3b90b6647679622c20ac96bb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:39:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7777ca58dcb544b3b8ef0fea36995e8a : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fd97fdd1008437daf715db81acff190 : SUCCESS in 3m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0487af07c86c429b84089947e69dc9fe : SUCCESS in 3m 27s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/629ae4e4924144dba0086dd6019ab4c8 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5b8f3fa217f19a4eea34f6d7e77b7c8763a898fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:40:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/b2d9317565fa447fa027b0b0e110aa4f : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a57b25c840ec26820ef53a49984754ab8568571c","revisions":{"a57b25c840ec26820ef53a49984754ab8568571c":{"kind":"REWORK","_number":1,"created":"2020-09-22 06:51:39.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/29/753229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/29/753229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/29/753229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/29/753229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/29/753229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/29/753229/1"}}},"commit":{"parents":[{"commit":"2416f0cf05e1fbd420c31aff710d590544a126f1","subject":"Correct TC Election event in template_emails.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2416f0cf05e1fbd420c31aff710d590544a126f1"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-18 12:18:00.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-22 06:51:24.000000000","tz":120},"subject":"Add Artem Goncharov candidacy for SDK PTL","message":"Add Artem Goncharov candidacy for SDK PTL\n\nChange-Id: I35dd6952f2ee038a9d7937dca87f02ca3423fc01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a57b25c840ec26820ef53a49984754ab8568571c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a57b25c840ec26820ef53a49984754ab8568571c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
