)]}'
{"id":"openstack%2Felection~753503","triplet_id":"openstack%2Felection~master~I89b31ca336d3ca96619fd0014c759b5f5344e5db","project":"openstack/election","branch":"master","hashtags":[],"change_id":"I89b31ca336d3ca96619fd0014c759b5f5344e5db","subject":"Add Marios Andreou candidacy for TripleO PTL","status":"MERGED","created":"2020-09-23 05:31:02.000000000","updated":"2020-09-24 07:40:36.000000000","submitted":"2020-09-23 19:41:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753503-1600890080712-2eec13d7","meta_rev_id":"e6875326ed425bb92207b7e72a03e6daf2025a9e","_number":753503,"virtual_id_number":753503,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-23 13:00:48.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"date":"2020-09-23 08:42:09.000000000","_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-23 19:41:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2020-09-23 13:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-23 19:30:44.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"date":"2020-09-23 14:16:02.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-23 19:30:44.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 08:42:09.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2020-09-23 13:47:51.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-09-23 14:16:02.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-09-23 19:30:44.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-09-23 19:41:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61af03a6f520bb48190de21417dd5abc6e93354d","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-23 05:31:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"120b802f1fb406ecc0ab392c9fb55a615ad05a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 05:35:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b9127f092449179ad813b1d84839a2 : FAILURE in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7d709da72f945748b58990e43297691 : SUCCESS in 3m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/39b8f48c6cf7481aba78fbcdfeab435e : SUCCESS in 3m 09s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/a49162668a95489f978eda350f4f83c6 : FAILURE in 2m 54s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/3b978e7efd084a65830c3cb277ff18f7 : FAILURE in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"978f5ff9c7e8d79a9e0c503d8376c8ed2be79b7c","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-23 06:05:31.000000000","message":"Patch Set 1: Code-Review-1\n\nElection officials cannot validate you as a Foundation member - would you check this?","accounts_in_message":[],"_revision_number":1},{"id":"28fe85bacf93c6210504d484538757f85bb27776","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-23 06:33:14.000000000","message":"Patch Set 1:\n\n\u003e Election officials cannot validate you as a Foundation member -\n \u003e would you check this?\n\nthanks for checking so quickly Ian.\n\nI am not sure why/how that happened ... I mean I had an openstack ID, I signed the OpenStack Individual Contributor License Agreement many years ago. \n\nI think it should be OK now https://www.openstack.org/community/members/profile/11672/marios-andreou","accounts_in_message":[],"_revision_number":1},{"id":"b0e64dc3c31b82f750d3eabb8237504249d0b6d5","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-23 06:38:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2d620e0e1f0d9473e8ac25a19fe45324afbdab8f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-23 06:40:08.000000000","message":"Patch Set 1:\n\noh i see it was done in the docs job :)\n\nhttps://zuul.opendev.org/t/openstack/build/68b9127f092449179ad813b1d84839a2\n\nHandler \u003cfunction build_lists at 0x7f7affaae5e0\u003e for event \u0027builder-inited\u0027 threw an exception (exception: No Openstack Foundation (OSF) data found for marios@redhat.com please check that marios@redhat.com is a valid email address and linked to a valid OSF account)\n\nlet\u0027s see if it finds my profile now?","accounts_in_message":[],"_revision_number":1},{"id":"29e01587c00983a6ccc8e2bc949876f242461485","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-23 06:45:31.000000000","message":"Patch Set 1:\n\nhmm i see the same at https://review.opendev.org/#/c/753229/ \n\nhttps://zuul.opendev.org/t/openstack/build/3101cdf9c57347e2b8e15861c0a17ea4 Handler \u003cfunction build_lists at 0x7f2f5ee985e0\u003e for event \u0027builder-inited\u0027 threw an exception (exception: No Openstack Foundation (OSF) data found for artem.goncharov@gmail.com please check that artem.goncharov@gmail.com is a valid email address and linked to a valid OSF account)\n\nbut it was green for v1 there i wonder is there something wrong with the job","accounts_in_message":[],"_revision_number":1},{"id":"56922cbc35f6b9f452a72946bb876972c3ae3291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 06:59:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0230d8c603b74586a9f8162277331314 : FAILURE in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b59e7d88a24643119cb8e2d1ed1ee85d : SUCCESS in 3m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8eac15c2b856449998c883c9f842c3ed : SUCCESS in 6m 04s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8485e48a7af44a6cae60f2aaccf6994c : FAILURE in 3m 22s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/9889d968460b45218f0a26c5424b3947 : FAILURE in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3ec3734fbf8a9239c805ce4c81800d6dc9046ef2","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-23 07:07:04.000000000","message":"Patch Set 1:\n\n\u003e Build failed (check pipeline).  For information on how to proceed,\n \u003e see\n \u003e https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n \u003e \n \u003e \n \u003e - openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0230d8c603b74586a9f8162277331314\n \u003e : FAILURE in 5m 00s\n \u003e - openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b59e7d88a24643119cb8e2d1ed1ee85d\n \u003e : SUCCESS in 3m 53s\n \u003e - openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8eac15c2b856449998c883c9f842c3ed\n \u003e : SUCCESS in 6m 04s\n \u003e - election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8485e48a7af44a6cae60f2aaccf6994c\n \u003e : FAILURE in 3m 22s\n \u003e - election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/9889d968460b45218f0a26c5424b3947\n \u003e : FAILURE in 6m 17s\n\n:/\n\nIan any idea how I can resolve that? Is my profile missing something that the job is trying to query \n\nhttps://2ab1d6bb900a3c51a4af-814491f7a8e968b50eeaf14b0c7f115f.ssl.cf2.rackcdn.com/753503/1/check/election-tox-ci-checks-review/8485e48/job-output.txt\n\n2020-09-23 06:51:42.660604 | ubuntu-focal | Validate email address is OSF member\n2020-09-23 06:51:42.660624 | ubuntu-focal | ------------------------------------\n2020-09-23 06:51:42.660642 | ubuntu-focal | Queried: https://openstackid-resources.openstack.org/api/public/v1/members?expand\u003dgroups\u0026filter%5B%5D\u003demail%3D%3Dmarios%40redhat.com\u0026filter%5B%5D\u003dgroup_slug%3D%3Dfoundation-members\u0026relations\u003daffiliations%2Cgroups","accounts_in_message":[],"_revision_number":1},{"id":"8e1a4405bb3305fa236ead8f85668ce35d6840de","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-23 07:12:52.000000000","message":"Patch Set 1:\n\n\u003e \u003e Build failed (check pipeline).  For information on how to\n \u003e proceed,\n \u003e \u003e see\n \u003e \u003e https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n \u003e \u003e\n \u003e \u003e\n \u003e \u003e - openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0230d8c603b74586a9f8162277331314\n \u003e \u003e : FAILURE in 5m 00s\n \u003e \u003e - openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b59e7d88a24643119cb8e2d1ed1ee85d\n \u003e \u003e : SUCCESS in 3m 53s\n \u003e \u003e - openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8eac15c2b856449998c883c9f842c3ed\n \u003e \u003e : SUCCESS in 6m 04s\n \u003e \u003e - election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8485e48a7af44a6cae60f2aaccf6994c\n \u003e \u003e : FAILURE in 3m 22s\n \u003e \u003e - election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/9889d968460b45218f0a26c5424b3947\n \u003e \u003e : FAILURE in 6m 17s\n \u003e \n \u003e :/\n \u003e \n \u003e Ian any idea how I can resolve that? Is my profile missing\n \u003e something that the job is trying to query\n \u003e \n \u003e https://2ab1d6bb900a3c51a4af-814491f7a8e968b50eeaf14b0c7f115f.ssl.cf2.rackcdn.com/753503/1/check/election-tox-ci-checks-review/8485e48/job-output.txt\n \u003e \n \u003e 2020-09-23 06:51:42.660604 | ubuntu-focal | Validate email address\n \u003e is OSF member\n \u003e 2020-09-23 06:51:42.660624 | ubuntu-focal | ------------------------------------\n \u003e 2020-09-23 06:51:42.660642 | ubuntu-focal | Queried:\n \u003e https://openstackid-resources.openstack.org/api/public/v1/members?expand\u003dgroups\u0026filter%5B%5D\u003demail%3D%3Dmarios%40redhat.com\u0026filter%5B%5D\u003dgroup_slug%3D%3Dfoundation-members\u0026relations\u003daffiliations%2Cgroups\n\nmy profile at https://www.openstack.org/profile/ keeps switching me back to community member... i have agreed to the license at least 3 times today already :D\n\nI have just clicked it again but I am not sure what is going on.","accounts_in_message":[],"_revision_number":1},{"id":"565523e59aa7facba6f7b20fe4f616921c4c64cc","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-23 07:14:36.000000000","message":"Patch Set 1:\n\nrecheck\n\nI *again* agreed to the license and the profile now again says \u0027foundation member\u0027 ... https://www.openstack.org/community/members/profile/11672/marios-andreou  \nif that gives you 404 then it switched my back again...","accounts_in_message":[],"_revision_number":1},{"id":"252eab26f883e7fc3811a346e8cb4411fdcaf3d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 07:31:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deea29814ee747f1b93107b9bd96561a : FAILURE in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5292eba37b9443fa5ead9c946309e2b : SUCCESS in 4m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fcecd3ff23e843eb90023437c6edb194 : SUCCESS in 6m 17s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/b4f584fe40814de69c774165f04e37bc : FAILURE in 4m 29s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/3f83d302fe4d4ecbb86ba5b2db28aa3d : FAILURE in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ead77985ea45735d9d5cca5c8682c2ed5a33b1f8","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-23 08:42:09.000000000","message":"Patch Set 1:\n\nevery change is failing this way and there is corresponding statement in the irc channel (http://eavesdrop.openstack.org/irclogs/%23openstack-election/%23openstack-election.2020-09-23.log.html). I hope somebody will take a look soon","accounts_in_message":[],"_revision_number":1},{"id":"ab612ca165d572d46cabcb74ba70cf8622c1cbf8","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-23 08:46:32.000000000","message":"Patch Set 1:\n\n\u003e every change is failing this way and there is corresponding\n \u003e statement in the irc channel (http://eavesdrop.openstack.org/irclogs/%23openstack-election/%23openstack-election.2020-09-23.log.html).\n \u003e I hope somebody will take a look soon\n\nthank you !","accounts_in_message":[],"_revision_number":1},{"id":"e696abfcd7fa13cb33c60dbebbf0b784096b6430","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-23 10:30:52.000000000","message":"Patch Set 1: -Code-Review\n\nRight thank u for sharing IRC conversation link. Election officials will review once the failing is sorted out\n\n \u003e \u003e every change is failing this way and there is corresponding\n \u003e \u003e statement in the irc channel (http://eavesdrop.openstack.org/irclogs/%23openstack-election/%23openstack-election.2020-09-23.log.html).\n \u003e \u003e I hope somebody will take a look soon\n \u003e \n \u003e thank you !","accounts_in_message":[],"_revision_number":1},{"id":"61ea3e43dbc5a97e6e9cd24494464434b3900336","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-23 12:56:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"efb570cdbf7e974632667ee08ff4285eda8ca751","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-23 13:00:48.000000000","message":"Patch Set 1:\n\n\u003e Right thank u for sharing IRC conversation link. Election officials\n \u003e will review once the failing is sorted out\n \u003e \n \u003e \u003e \u003e every change is failing this way and there is corresponding\n \u003e \u003e \u003e statement in the irc channel (http://eavesdrop.openstack.org/irclogs/%23openstack-election/%23openstack-election.2020-09-23.log.html).\n \u003e \u003e \u003e I hope somebody will take a look soon\n \u003e \u003e\n \u003e \u003e thank you !\n\n \u003e recheck\n\nthank you for checking again Ian I see that is now returning a result so hopeful :) \n\nhttps://openstackid-resources.openstack.org/api/public/v1/members?expand\u003dgroups\u0026filter%5B%5D\u003demail%3D%3Dmarios%40redhat.com\u0026filter%5B%5D\u003dgroup_slug%3D%3Dfoundation-members\u0026relations\u003daffiliations%2Cgroups","accounts_in_message":[],"_revision_number":1},{"id":"a0a80331acf5ebbd895a468b520b8583a8aa7f7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 13:40:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97257971bfb94036b79e622df3a09571 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b35e89b03cff4c02ac654e51b081d6f7 : SUCCESS in 3m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/31dc126bacb840848612e9ba2cb5c08e : SUCCESS in 3m 07s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/02c8a43fed3c421b8d371df8f445dbae : SUCCESS in 3m 22s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5aabcc761ba54adfa91c7aee120175d9 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c08a39bf142393ec5e7cc728917c60f41bec798f","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-23 13:47:51.000000000","message":"Patch Set 1: Code-Review+2\n\nConfirms as a valid candidate","accounts_in_message":[],"_revision_number":1},{"id":"6ee1448162c91333440a92167ce66dc745b20ce7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-23 19:30:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nhttps://openstack.org/community/members/profile/11672\nhttps://review.opendev.org/696874 merged on 2019-12-06 to openstack/paunch for tripleo","accounts_in_message":[],"_revision_number":1},{"id":"c0a5fbf2c72777ecb0de186c1b52dcc3a2d74eb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:30:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"972e38d46f74ec31f0cc2b259baca6aae13abdce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:41:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5b4a0464b1f1ae84ec06190bdc5a6e5c87d01b0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:41:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427f1f22839a48e98f728115034c918c : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4310ee96b2594f9b948afca92b7c2f90 : SUCCESS in 3m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fbc632564749456192339e024ce2ede6 : SUCCESS in 3m 26s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/80373cab6a2047eeb11ea02b78e3a5c2 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"115af2f9083a1f28b6dab0936238c304f5ece234","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:42:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/217ea6f5c1f64dcf805c82a442042906 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1},{"id":"481eaef0d0e7394392cab8bdc44dc4612d892fd6","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-24 05:21:52.000000000","message":"Patch Set 1:\n\n\u003e https://openstack.org/community/members/profile/11672\n \u003e https://review.opendev.org/696874 merged on 2019-12-06 to\n \u003e openstack/paunch for tripleo\n\njust wondering what is the relevance of that particular patch it\u0027s quite old :)","accounts_in_message":[],"_revision_number":1},{"id":"571eec02ab8c45d669849b3fe391df571594e5bf","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-24 05:24:38.000000000","message":"Patch Set 1:\n\n\u003e \u003e https://openstack.org/community/members/profile/11672\n \u003e \u003e https://review.opendev.org/696874 merged on 2019-12-06 to\n \u003e \u003e openstack/paunch for tripleo\n \u003e \n \u003e just wondering what is the relevance of that particular patch it\u0027s\n \u003e quite old :)\n\nI guess you were ensuring I made a commit to one of the tripleo repos \n\nhttps://review.opendev.org/#/q/status:merged+owner:marios@redhat.com","accounts_in_message":[],"_revision_number":1},{"id":"e6875326ed425bb92207b7e72a03e6daf2025a9e","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-09-24 07:40:36.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e https://openstack.org/community/members/profile/11672\n \u003e \u003e \u003e https://review.opendev.org/696874 merged on 2019-12-06 to\n \u003e \u003e \u003e openstack/paunch for tripleo\n \u003e \u003e\n \u003e \u003e just wondering what is the relevance of that particular patch\n \u003e it\u0027s\n \u003e \u003e quite old :)\n \u003e \n \u003e I guess you were ensuring I made a commit to one of the tripleo\n \u003e repos\n \u003e \n \u003e https://review.opendev.org/#/q/status:merged+owner:marios@redhat.com\n\nso it must be you were looking for merges between Sep 27, 2019 00:00 UTC to Sep 29, 2020 00:00 UTC","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03ffc895d37801bf7e505d0d5a4d4986ccf02e40","revisions":{"03ffc895d37801bf7e505d0d5a4d4986ccf02e40":{"kind":"REWORK","_number":1,"created":"2020-09-23 05:31:02.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/03/753503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/03/753503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/03/753503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/03/753503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/03/753503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/03/753503/1"}}},"commit":{"parents":[{"commit":"2416f0cf05e1fbd420c31aff710d590544a126f1","subject":"Correct TC Election event in template_emails.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2416f0cf05e1fbd420c31aff710d590544a126f1"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2020-09-23 05:29:43.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2020-09-23 05:29:43.000000000","tz":180},"subject":"Add Marios Andreou candidacy for TripleO PTL","message":"Add Marios Andreou candidacy for TripleO PTL\n\nChange-Id: I89b31ca336d3ca96619fd0014c759b5f5344e5db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/03ffc895d37801bf7e505d0d5a4d4986ccf02e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/03ffc895d37801bf7e505d0d5a4d4986ccf02e40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
