)]}'
{"id":"openstack%2Felection~753862","triplet_id":"openstack%2Felection~master~I6cf84cb467e7fe78ecf50395287e06e9904c5bdd","project":"openstack/election","branch":"master","topic":"tc-candidacy","hashtags":[],"change_id":"I6cf84cb467e7fe78ecf50395287e06e9904c5bdd","subject":"Add TC Candidacy for Kendall Nelson","status":"MERGED","created":"2020-09-23 23:02:54.000000000","updated":"2020-09-24 16:19:49.000000000","submitted":"2020-09-24 16:18:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753862-1600964299296-60701959","meta_rev_id":"344bcb1e2247beb637c72f4d34740112f71ce636","_number":753862,"virtual_id_number":753862,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 16:18:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2020-09-24 07:41:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-24 16:05:30.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-24 16:05:30.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 07:41:45.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-09-24 16:05:30.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-09-24 16:18:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c3faeb3ea745ddc4d1bde9c7520c197613d92cd","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-09-23 23:02:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a86a3d11017f8ac7d7243c9a9820a433e70c3034","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-09-23 23:04:03.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"66d7f8a286db5e9543ccdc1ba4024eb462d1cb48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 23:09:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b718c24e7f2e471089129353065a072a : FAILURE in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b42874bb02e4a96a46acaf151742fd7 : SUCCESS in 3m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/84cd1e03379f4520a74060d0f6b26170 : SUCCESS in 3m 31s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/b7653d94eaa24e78bd2f8e0d3ca1fc8c : SUCCESS in 3m 21s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/6dcb3ef80b9e4c3688b068b5b9ab5a9b : FAILURE in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"3bd45cb3bf1b94b97b09890ffde0f4d177112602","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-23 23:57:20.000000000","message":"Patch Set 2:\n\nSeems that the job fails due to other\u0027s validity - will check more details.. [1][2]\n\n[1] https://zuul.opendev.org/t/openstack/build/6dcb3ef80b9e4c3688b068b5b9ab5a9b/log/job-output.txt#635\n[2]  https://review.opendev.org/#/c/753562/","accounts_in_message":[],"_revision_number":2},{"id":"24312a6e809c2988c4728fbcdbf11225e1956414","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-24 02:57:32.000000000","message":"Patch Set 2:\n\nYes, apparently candidates/wallaby/Kuryr/maysa.macedo95@gmail.com has suddenly started failing the OSF membership check. Manual checking indicates that the E-mail address is still associated with an OSF profile but that the profile is no longer for an Individual Member of the OSF. It\u0027s possible they accidentally resigned their membership, in which case we\u0027ll need to revert their candidacy until it can be corrected.","accounts_in_message":[],"_revision_number":2},{"id":"e2db7b75e3fda72bd35d1b3d69c29f16d0788742","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-24 03:58:32.000000000","message":"Patch Set 2:\n\nhttps://review.opendev.org/753924 should hopefully get things moving again.","accounts_in_message":[],"_revision_number":2},{"id":"88f92ec10d865e85e41b276e9398a122f4985c71","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-24 06:53:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4aa7ad4f2efbcef26d4d0d411f88419a7da627fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 07:10:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e8968d71ee4082983f96840e684bb4 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5306a38422cd4eda88b4eb7018a37b8e : SUCCESS in 4m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/04dcbf1a8ec14211b581b7988c8798e3 : SUCCESS in 4m 30s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/cabcc717fef14073b58b91751a61a603 : SUCCESS in 5m 10s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/382e05bf6aa84ace95304f32db1b06b2 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"8a9f7a7c34a8acca38e8baec8ab7385faad60c82","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-24 07:41:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"81e5e0874c27e5c5a6f59a82bf9800bd449a85dd","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-24 07:45:03.000000000","message":"Patch Set 2:\n\nConfirms as a valid tc candidate","accounts_in_message":[],"_revision_number":2},{"id":"9681b129a85d67570c4e94b33cd6937dec426e15","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-24 16:05:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nhttps://openstack.org/community/members/profile/35859","accounts_in_message":[],"_revision_number":2},{"id":"ec86378ad55eb42cb9ef8b99060f99538da7c018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:05:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0af5f252509fa2e4335d841af9ff1fa2dbbbbc83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:18:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7f1f7b810d7e42250b0ee4b2f0058d0a1d2171a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:18:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db5ec073e054eceb99f73c81795c655 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89a8e64f5be444b5b25d3f2cb17a36a4 : SUCCESS in 5m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b018c9eb53174005b791d13cdc74adef : SUCCESS in 4m 26s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c631939ee8494e868ff1bf860abfd610 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5bf2e0b1539a8f13606a37743a5e433346c080d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:19:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/ebafd0d29fb7490487819caea4e265d2 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f4314feeb616a053433e08207766837fd8a3c33","revisions":{"f8eb185c7aa5d6415ace758554817fea7aa82585":{"kind":"REWORK","_number":1,"created":"2020-09-23 23:02:54.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/62/753862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/62/753862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/62/753862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/62/753862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/62/753862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/62/753862/1"}}},"commit":{"parents":[{"commit":"b5750d843e052c7a3cb459843fec992679967ec2","subject":"Merge \"Adding Balazs Gibizer candidacy for Nova PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b5750d843e052c7a3cb459843fec992679967ec2"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-09-23 23:02:26.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-09-23 23:02:26.000000000","tz":-420},"subject":"Add TC Candidacy for Kendall Nelson","message":"Add TC Candidacy for Kendall Nelson\n\nChange-Id: I6cf84cb467e7fe78ecf50395287e06e9904c5bdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f8eb185c7aa5d6415ace758554817fea7aa82585"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f8eb185c7aa5d6415ace758554817fea7aa82585"}]},"branch":"refs/heads/master"},"3f4314feeb616a053433e08207766837fd8a3c33":{"kind":"REWORK","_number":2,"created":"2020-09-23 23:04:03.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/62/753862/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/62/753862/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/62/753862/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/62/753862/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/62/753862/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/62/753862/2"}}},"commit":{"parents":[{"commit":"b5750d843e052c7a3cb459843fec992679967ec2","subject":"Merge \"Adding Balazs Gibizer candidacy for Nova PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b5750d843e052c7a3cb459843fec992679967ec2"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-09-23 23:02:26.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-09-23 23:03:58.000000000","tz":0},"subject":"Add TC Candidacy for Kendall Nelson","message":"Add TC Candidacy for Kendall Nelson\n\nChange-Id: I6cf84cb467e7fe78ecf50395287e06e9904c5bdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3f4314feeb616a053433e08207766837fd8a3c33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3f4314feeb616a053433e08207766837fd8a3c33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
