)]}'
{"id":"openstack%2Felection~753924","triplet_id":"openstack%2Felection~master~I98f7575f12cf193a03f1126668c0dacc2998ee70","project":"openstack/election","branch":"master","topic":"wallaby","hashtags":[],"change_id":"I98f7575f12cf193a03f1126668c0dacc2998ee70","subject":"Revert \"Adding Maysa de Macedo Souza candidacy for Kuryr\"","status":"ABANDONED","created":"2020-09-24 03:07:36.000000000","updated":"2020-09-24 12:38:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9b698ea57aab6d0d4a3ac0f53d3ee88730f7ef47","_number":753924,"virtual_id_number":753924,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"date":"2020-09-24 07:04:50.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-24 04:04:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":-1,"date":"2020-09-24 07:43:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 04:04:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-24 07:04:50.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2020-09-24 07:43:18.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"135073a70a34205b87063a8e97ff28af55b42471","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-24 03:07:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59cab70ccc375defdfcf205d9a6df6dbbb89e66e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 03:14:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da256ef6a89d44d3a0b78749830de4cf : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9d2432830b14130936eeda951fca2ba : SUCCESS in 3m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9bfc2f36d13b443294d2996b2bca9c0e : SUCCESS in 3m 43s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/b5f2fc2844a04d8e818c39bbe95ea457 : FAILURE in 3m 18s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/bac0d90026ba464ea0de4e88a02580e9 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"28a6270e0f405692d2c8d22d2d01b3b6efba71b5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-24 03:42:05.000000000","message":"Patch Set 1:\n\nUnfortunately the election-tox-ci-checks-review job handles reverts in an unexpected manner. It seems to pick up the address being removed in the diff and tests it as it would an addition. We\u0027ll either need to bypass the job temporarily or fix it (and of course we should fix it either way regardless).","accounts_in_message":[],"_revision_number":1},{"id":"c36b532f973a2b80bfe66a4aaf746cb630e00a2e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-24 03:56:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f8ead914f2aa41d8b06ae15b862190d9b118cb63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 04:04:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f755b47059943ffa87701505ce8c133 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e98adca97cf942dd8b8bc06b74d4dd37 : FAILURE in 3m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a7162259c594473aa97de908c369928 : SUCCESS in 5m 37s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/3ef95e0dd2274001bf5b468759bfe6cb : SUCCESS in 2m 56s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/852ccf1377c244f1a5f6c3e9174fa089 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"254d6dda93ca44a0928aa06083573799b4fac296","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-24 04:22:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(Fast-approval due to merging issues)","accounts_in_message":[],"_revision_number":2},{"id":"d6288b110bbdbc4dac3c91d693494a1ff4cef837","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-24 04:24:36.000000000","message":"Patch Set 2: -Code-Review -Workflow\n\n(Reverting +A - need to see in details..)","accounts_in_message":[],"_revision_number":2},{"id":"e4fe33e33f70b1f150a41ee0509a004b2fa1bbe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 04:32:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0be0c068a3f94747ab05582f4a002ad3 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2e9ea5c3c7d4a28b1f4041f97de17b6 : FAILURE in 3m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ad62f704a6444d2799e0320e59a3a496 : SUCCESS in 3m 28s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/5899037c77814376b5534320e15b686c : SUCCESS in 5m 56s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/67e2a3df2c8d49a3b79b27db6e516e6d : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d8618db9c45bd93f0c18079fbc858d4bcbf91119","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2020-09-24 07:04:50.000000000","message":"Patch Set 2:\n\nThe issue seems to be fixed now. The only thing I did before was update the description on my profile, and somehow that removed me from being a Foundation member.","accounts_in_message":[],"_revision_number":2},{"id":"8a45e706e691b2c985940f08e0586b591c6d4fe0","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-24 07:43:18.000000000","message":"Patch Set 2: Workflow-1\n\nNo need to revert since the candidate checked OSF status and now it is fine","accounts_in_message":[],"_revision_number":2},{"id":"ef1e9193c7271546ad4cfb1afd343b87bc19aa5f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-24 12:38:28.000000000","message":"Abandoned\n\nProblem resolved, thanks!","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"86d8f3cc087ecfbb88650d99f9c147e38adba0c9","revisions":{"34a9f6737441ac2eb43b46da9caf46072db3f26a":{"kind":"REWORK","_number":1,"created":"2020-09-24 03:07:36.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/24/753924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/24/753924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/24/753924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/24/753924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/24/753924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/24/753924/1"}}},"commit":{"parents":[{"commit":"b5750d843e052c7a3cb459843fec992679967ec2","subject":"Merge \"Adding Balazs Gibizer candidacy for Nova PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b5750d843e052c7a3cb459843fec992679967ec2"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-09-24 03:06:24.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-09-24 03:07:19.000000000","tz":0},"subject":"Revert \"Adding Maysa de Macedo Souza candidacy for Kuryr\"","message":"Revert \"Adding Maysa de Macedo Souza candidacy for Kuryr\"\n\nThis reverts commit ae5136154c09a5dded0e16e8e817e2d041d88933.\n\nThe www.openstack.org account for this address no longer corresponds\nto an OSF Individual Membership. In the short term, we\u0027re going to\nrevert this change so we can continue to process other nominees. The\ncandidacy should be resubmitted once you have reinstated your OSF\nIndividual Membership.\n\nChange-Id: I98f7575f12cf193a03f1126668c0dacc2998ee70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/34a9f6737441ac2eb43b46da9caf46072db3f26a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/34a9f6737441ac2eb43b46da9caf46072db3f26a"}]},"branch":"refs/heads/master"},"86d8f3cc087ecfbb88650d99f9c147e38adba0c9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-24 03:56:06.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/24/753924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/24/753924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/24/753924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/24/753924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/24/753924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/24/753924/2"}}},"commit":{"parents":[{"commit":"86d143609a73d6b899e63ce20a3bca95279a174e","subject":"Don\u0027t test candidate files if they\u0027re removed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/86d143609a73d6b899e63ce20a3bca95279a174e"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-09-24 03:06:24.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-09-24 03:55:47.000000000","tz":0},"subject":"Revert \"Adding Maysa de Macedo Souza candidacy for Kuryr\"","message":"Revert \"Adding Maysa de Macedo Souza candidacy for Kuryr\"\n\nThis reverts commit ae5136154c09a5dded0e16e8e817e2d041d88933.\n\nThe www.openstack.org account for this address no longer corresponds\nto an OSF Individual Membership. In the short term, we\u0027re going to\nrevert this change so we can continue to process other nominees. The\ncandidacy should be resubmitted once you have reinstated your OSF\nIndividual Membership.\n\nChange-Id: I98f7575f12cf193a03f1126668c0dacc2998ee70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/86d8f3cc087ecfbb88650d99f9c147e38adba0c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/86d8f3cc087ecfbb88650d99f9c147e38adba0c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
