)]}'
{"id":"openstack%2Felection~754214","triplet_id":"openstack%2Felection~master~Iffc84aeaad3958f16a070bf3d7cc3b3ac683ccb3","project":"openstack/election","branch":"master","hashtags":[],"change_id":"Iffc84aeaad3958f16a070bf3d7cc3b3ac683ccb3","subject":"Add Slawek Kaplonski candidacy for Neutron PTL","status":"MERGED","created":"2020-09-24 20:37:53.000000000","updated":"2020-09-25 00:28:57.000000000","submitted":"2020-09-25 00:27:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754214-1600993647331-5dd22bf2","meta_rev_id":"363f9c21215330bdaaaae89155b774e3c08326d3","_number":754214,"virtual_id_number":754214,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 00:27:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2020-09-25 00:21:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-24 22:21:07.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2020-09-25 00:21:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 22:21:07.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-09-25 00:21:20.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-09-25 00:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5cb6056826fbf447766958ee892db47a46127b87","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-24 20:37:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccefec1a815162d3c319e63b943471b58d7cb382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 20:48:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5afc18fb930446cb29518518ce8b5e7 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1781169e0af34830b4a1c894d5b0094c : SUCCESS in 3m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/906e55d8e4e6440588830636447bdb96 : SUCCESS in 3m 32s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/0177a09e8ed74309a28e7bb31f4160eb : SUCCESS in 3m 12s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/0cd00db7cd8d40c6a62a7d7dd7d604b1 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"59a6391f23de94f3a529ff4f39dad90d70699b32","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-24 22:21:07.000000000","message":"Patch Set 1: Code-Review+2\n\nhttps://openstack.org/community/members/profile/19377\nhttps://review.opendev.org/726782 merged on 2020-05-18 to openstack/networking-bagpipe for neutron","accounts_in_message":[],"_revision_number":1},{"id":"490c3e89f1f1c2cc6df3f3925b6d0125c759b5fc","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-25 00:21:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aff7e79dcec97da0cb8144cdd3a8c7c8c787b2f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 00:21:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b49d75fce4271d4fb6c36faa2900bdb9ee508de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 00:27:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5dda8216e6b8a4afe17acedd5d75c8fc2fefdb35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 00:27:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99fecc4d505f482e884e5aaa3894a4ef : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/039ffbf5d9ee4848bedd1d4936072ad3 : SUCCESS in 5m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5bbcd05032134b3688c2900264cef767 : SUCCESS in 3m 14s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/69c189b51d7e4ad2b53a6a2722ab1973 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"363f9c21215330bdaaaae89155b774e3c08326d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 00:28:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/4cf580670b704141b7b0dbb88a0ef168 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ced039b021c7443b2cc481d2a601672cdc856334","revisions":{"ced039b021c7443b2cc481d2a601672cdc856334":{"kind":"REWORK","_number":1,"created":"2020-09-24 20:37:53.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/14/754214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/14/754214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/14/754214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/14/754214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/14/754214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/14/754214/1"}}},"commit":{"parents":[{"commit":"ea76f7aba26dd8f6244c9eb5138f3b1771fdeca7","subject":"Merge \"Add Lance Albertson candidancy for Openstack Chef PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ea76f7aba26dd8f6244c9eb5138f3b1771fdeca7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-24 20:37:02.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-09-24 20:37:48.000000000","tz":120},"subject":"Add Slawek Kaplonski candidacy for Neutron PTL","message":"Add Slawek Kaplonski candidacy for Neutron PTL\n\nChange-Id: Iffc84aeaad3958f16a070bf3d7cc3b3ac683ccb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ced039b021c7443b2cc481d2a601672cdc856334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ced039b021c7443b2cc481d2a601672cdc856334"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
