)]}'
{"id":"openstack%2Felection~754617","triplet_id":"openstack%2Felection~master~If89b7c7e236fa340de0d5e78740e48b0e3f28199","project":"openstack/election","branch":"master","topic":"wallaby-trove-ptl","hashtags":[],"change_id":"If89b7c7e236fa340de0d5e78740e48b0e3f28199","subject":"Adding Lingxian Kong candidacy for Trove","status":"MERGED","created":"2020-09-27 21:05:23.000000000","updated":"2020-09-28 01:59:08.000000000","submitted":"2020-09-28 01:58:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754617-1601258283375-e5a30d7c","meta_rev_id":"50f689a91570ac313ef945fac8e7bcbe0122beab","_number":754617,"virtual_id_number":754617,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-28 01:58:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2020-09-28 01:47:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-27 21:17:58.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2020-09-28 01:47:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-27 21:17:58.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-09-28 01:47:15.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-09-28 01:58:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b90e00a275e37f4fcb9d2eb05e6268bacff12b64","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-27 21:05:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32abeb6c79d8ff60393edb48eaabe09f64428bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 21:11:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6879aedc8c548bf9232e576808ffc9c : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbc25a5c0fbc40df81d6b78e79822482 : SUCCESS in 3m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5e45ab9c7406468d811af7d7754bb3c0 : SUCCESS in 3m 39s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/c903b371718649df97d9883aa3129871 : SUCCESS in 2m 53s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/add9256585ef4a1faef7a6f638256ff2 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9edf0d4d87bc25e6e27d4123d86f0463dc594ae6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-27 21:17:58.000000000","message":"Patch Set 1: Code-Review+2\n\nhttps://openstack.org/community/members/profile/21438\nhttps://review.opendev.org/751913 merged on 2020-09-15 to openstack/python-troveclient for trove","accounts_in_message":[],"_revision_number":1},{"id":"70506b3a907f306cad3fc9a84a9ee0d6f79e27ef","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-28 01:47:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nconfirms as a valid candidate","accounts_in_message":[],"_revision_number":1},{"id":"b1bf504ab171bc1043ea1ae8a4ba9f9dce4d48c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 01:47:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf01d7f47643f29ccebb692cace50f605e2ee33f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 01:58:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"978a7d4e6750a061f3e91fbe5557aad0f0c6d5e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 01:58:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55962d9300494122b0d49f631a23891c : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a7913abf2224c0f94ef3e4981d300c9 : SUCCESS in 3m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/16f62492160e47f4b415404403b8a60c : SUCCESS in 2m 54s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/57208bb59bc94b42a1b38c1c3dc7b6d3 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f4d13fae8e3b6e327f2a2dfd7138ebb920e624b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 01:59:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/a954ed6d83524d20b527ab17542cfd74 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac6f212d6ead9c4247e4651dc147bfe26c2106bd","revisions":{"ac6f212d6ead9c4247e4651dc147bfe26c2106bd":{"kind":"REWORK","_number":1,"created":"2020-09-27 21:05:23.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/17/754617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/17/754617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/17/754617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/17/754617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/17/754617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/17/754617/1"}}},"commit":{"parents":[{"commit":"3807f239acca889f0986dfa538009fb01db6a7a8","subject":"Merge \"Add Shengping Zhong candidacy for Puppet OpenStack PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3807f239acca889f0986dfa538009fb01db6a7a8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-27 21:03:15.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-27 21:03:15.000000000","tz":780},"subject":"Adding Lingxian Kong candidacy for Trove","message":"Adding Lingxian Kong candidacy for Trove\n\nChange-Id: If89b7c7e236fa340de0d5e78740e48b0e3f28199\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ac6f212d6ead9c4247e4651dc147bfe26c2106bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ac6f212d6ead9c4247e4651dc147bfe26c2106bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
