)]}'
{"id":"openstack%2Felection~754981","triplet_id":"openstack%2Felection~master~I2f129c72e5a532f753227dd7b14e55c532184583","project":"openstack/election","branch":"master","topic":"horizon-w","hashtags":[],"change_id":"I2f129c72e5a532f753227dd7b14e55c532184583","subject":"Add Ivan Kolodyazhny candidacy for Horizon PTL","status":"MERGED","created":"2020-09-29 12:04:15.000000000","updated":"2020-09-29 22:58:38.000000000","submitted":"2020-09-29 22:57:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754981-1601420243949-276e733d","meta_rev_id":"493952c1a9f86c80e72cbb6523f526c02683614b","_number":754981,"virtual_id_number":754981,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-29 22:57:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2020-09-29 22:49:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-29 14:54:42.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2020-09-29 22:49:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 14:54:42.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-09-29 22:49:12.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-09-29 22:57:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0626374a74e015bc251fa84a8367b45305acbe74","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-29 12:04:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93ee164a7c4161fab46f1ea5e3b13f7b7d81ad6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 12:10:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa89f88c9234b5c8e13f3213e1e5b2d : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/382c2b8c134d4f7886596c992bba970a : SUCCESS in 3m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e643cdc76bbe493fb797c5909f2d3c84 : SUCCESS in 3m 28s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/782b97124bb34339b66bf2a46ac456c1 : SUCCESS in 3m 22s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e78b54fbece54291a06f9025e85e55db : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a05b208b2412ea3ff7e4c67c83c7ff0ea993c0f2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-29 14:54:42.000000000","message":"Patch Set 1: Code-Review+2\n\nhttps://openstack.org/community/members/profile/3706\nhttps://review.opendev.org/721992 merged on 2020-09-24 to openstack/horizon for horizon","accounts_in_message":[],"_revision_number":1},{"id":"b920515335728d072d4cad6274840ec7858f41d5","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-29 22:49:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"05fb782bc634bb5890fcc3789b614482683556e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:49:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"190802bb147db958fa9d809671d623c63b1c34d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:57:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"efde4f7b83b928b46207ad0fa599fbe97f8b7f5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:57:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c85056a512954f6f8e98695053b0441f : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3920b21fca345aa9815d47bc3b2ca01 : SUCCESS in 5m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f92d43abdfe8492eba9ac595c65d8d14 : SUCCESS in 5m 24s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/bc3f010565d04401b02d9473e26f2abf : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0fcf9fbb232702d8f33784fbebcc9b13d824b44a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:58:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/6e4d1d64fbc241b89e4d9e5408bbfeca : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c100584c7b5620925f0835b980fd9b51edd4a67","revisions":{"3c100584c7b5620925f0835b980fd9b51edd4a67":{"kind":"REWORK","_number":1,"created":"2020-09-29 12:04:15.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/81/754981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/754981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/754981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/754981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/754981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/754981/1"}}},"commit":{"parents":[{"commit":"97871aed767a3f9454a6071ab9846a6879cde440","subject":"Merge \"Adding Takashi Kajinami candidacy for Storlets PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/97871aed767a3f9454a6071ab9846a6879cde440"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-29 12:03:30.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-29 12:03:30.000000000","tz":180},"subject":"Add Ivan Kolodyazhny candidacy for Horizon PTL","message":"Add Ivan Kolodyazhny candidacy for Horizon PTL\n\nChange-Id: I2f129c72e5a532f753227dd7b14e55c532184583\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3c100584c7b5620925f0835b980fd9b51edd4a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3c100584c7b5620925f0835b980fd9b51edd4a67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
