)]}'
{"id":"openstack%2Felection~755063","triplet_id":"openstack%2Felection~master~Ic880e0d313ba829e175a9c1701edf185061db0fc","project":"openstack/election","branch":"master","hashtags":[],"change_id":"Ic880e0d313ba829e175a9c1701edf185061db0fc","subject":"Add mrunge candidacy for Telemetry PTL","status":"MERGED","created":"2020-09-29 15:26:01.000000000","updated":"2020-09-29 23:09:58.000000000","submitted":"2020-09-29 23:08:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755063-1601420923908-398603b0","meta_rev_id":"bcfe9ae911de1f1795210425a77482bf6fd96641","_number":755063,"virtual_id_number":755063,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"date":"2020-09-29 15:28:00.000000000","_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-29 23:08:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2020-09-29 22:56:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-29 15:50:41.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2020-09-29 22:56:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 15:50:41.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-09-29 22:56:13.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-09-29 23:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1fad8efbcb13e0c8e22430d69612078ef764c790","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-09-29 15:26:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24b08ccd97bede7c9b4b3fe3dbd3ed6fccbf4e3c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-09-29 15:28:00.000000000","message":"Patch Set 1:\n\ncontributions to ceilometer, e.g. https://review.opendev.org/#/c/742089/","accounts_in_message":[],"_revision_number":1},{"id":"454e0aafcbe8897d7b775cfd6eba0741d35bdf75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 15:46:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae86a7aa82e44aabb8b20862430fd85 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0eeb005e02f04305a518a68936404bf1 : SUCCESS in 3m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4d855c8a7fce466ab0e3cf5770146ed3 : SUCCESS in 3m 53s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/04334a3bdc614e9691d45dbd5449b04e : SUCCESS in 6m 55s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c158aaf3e23f40178b08002a2cd1369e : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4f9b88a15f0818fc4cc8df299b9bf54918668b50","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-29 15:50:41.000000000","message":"Patch Set 1: Code-Review+2\n\nhttps://openstack.org/community/members/profile/5990\nhttps://review.opendev.org/743225 merged on 2020-07-28 to openstack/ceilometer for Telemetry","accounts_in_message":[],"_revision_number":1},{"id":"7c57c709e09cedd0257283754f8e540d9ac1f87e","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-29 22:56:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nConfirms as a valid candidate","accounts_in_message":[],"_revision_number":1},{"id":"d5668ab7b79b38cc9dacf2d7222443cf95b953d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:56:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9cc5d3220f3154e1f64a7ef933039aa664742d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 23:08:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8baf41b1701c466db7e8c496eb8e53be : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9564a81a81bf4129add8fd64dafba0da : SUCCESS in 3m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/43379e8b3f8a4ac8915b45786e65e54b : SUCCESS in 3m 42s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/22863979924b458c981ec5de6ad7f1a1 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"27c0c275f3ccc8d3b2c04fbcaa24428c1c9045ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 23:08:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bcfe9ae911de1f1795210425a77482bf6fd96641","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 23:09:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/e1c215a5af8d48499043cf348a1573a9 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49978bcefea95e93629b01db99301dff2448a461","revisions":{"49978bcefea95e93629b01db99301dff2448a461":{"kind":"REWORK","_number":1,"created":"2020-09-29 15:26:01.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/63/755063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/63/755063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/63/755063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/63/755063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/63/755063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/63/755063/1"}}},"commit":{"parents":[{"commit":"97871aed767a3f9454a6071ab9846a6879cde440","subject":"Merge \"Adding Takashi Kajinami candidacy for Storlets PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/97871aed767a3f9454a6071ab9846a6879cde440"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-09-29 15:22:30.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-09-29 15:25:58.000000000","tz":120},"subject":"Add mrunge candidacy for Telemetry PTL","message":"Add mrunge candidacy for Telemetry PTL\n\nfor Wallaby cycle\n\nChange-Id: Ic880e0d313ba829e175a9c1701edf185061db0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/49978bcefea95e93629b01db99301dff2448a461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/49978bcefea95e93629b01db99301dff2448a461"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
