)]}'
{"id":"openstack%2Felection~755141","triplet_id":"openstack%2Felection~master~Icfbff4ec284f227d2c575f43cf97e27190bca62e","project":"openstack/election","branch":"master","topic":"combined-templates","hashtags":[],"change_id":"Icfbff4ec284f227d2c575f43cf97e27190bca62e","subject":"Add a combined_end_nominations E-mail template","status":"MERGED","created":"2020-09-30 00:11:36.000000000","updated":"2021-01-15 20:46:23.000000000","submitted":"2021-01-15 20:46:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755141-combined-templates","meta_rev_id":"887258363ef23460a9d3a735e1ed775247e64563","_number":755141,"virtual_id_number":755141,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 20:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2020-10-29 12:22:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2021-01-15 20:34:02.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2021-01-15 20:35:18.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 21:35:32.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-10-26 15:08:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-29 12:22:45.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"6f7238f63390eba151f810c141a8ebf8e4dd1139","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-30 00:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbd3faa6fc31838dd17b4ac52daf748e750af89b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 00:21:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/851b46a1020047668bed7c9d5ef8bfaf : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5dafc49dc464f028cb1944b4b8a9dff : SUCCESS in 3m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0de8cef5a72141a69a46bd4bf479897e : SUCCESS in 3m 09s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/fcc4ef431776430984ff957b8d1a6b54 : SUCCESS in 3m 12s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/21e5355c5897457f8161a0849f95fda5 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8378c50b09ed2def0539b9f48942b780c3f08e7e","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-10-05 14:28:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)\n\nLooks great, and adding some comments from last combined_end_nominations template on Etherpad: https://etherpad.opendev.org/p/combined-nominations-close","accounts_in_message":[],"_revision_number":1},{"id":"fa60dc86276e97feaa42fae277b3ab5a3de5bf7b","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-10-26 14:46:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"773f27f3ccc5b33db9f140adc6d6d318580f52fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 15:08:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66b57cc74cd14fae8ae008432991e4ab : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e609250be2448c69f0169097d363f43 : SUCCESS in 8m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa459adf89fc428999d1a2274a235a94 : SUCCESS in 3m 06s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/3d74a5a0621344e3bab2771a037eecbe : SUCCESS in 3m 36s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/4447a85c6a7e40c18533741608982c3f : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"71d066a3fc3dc671fcbb980d047d112bded9ce80","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-10-29 12:22:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"497fddfa4cb73f6e7d3aedd5d0e1fc974c8347c3","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-01-15 20:34:02.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for the fix and the addition of the end nominations email!","accounts_in_message":[],"_revision_number":2},{"id":"a3a1f664c633f3f7e49836f9be9edd5a2a38b13e","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-01-15 20:35:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e93c0d202034f74e473de20755a0e2410f93a65e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:35:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5cf99e08c3f6b32adee3382cc259c7218173c093","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:46:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b314a7c1cfc349e49c424bcf4c6143fd : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91d3e20c9c0745ec9309a70d846c30d5 : SUCCESS in 3m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5601cad1d1724398993bca22f07f7bb5 : SUCCESS in 3m 21s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/050308d5d2a741069ad5861737e6d4b6 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"887258363ef23460a9d3a735e1ed775247e64563","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:46:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"119165be672d57aad4bbbadff8b3e6329feb491a","revisions":{"8536c504ef7c443921411e8dcf102dbc8381a871":{"kind":"REWORK","_number":1,"created":"2020-09-30 00:11:36.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/41/755141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/41/755141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/41/755141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/41/755141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/41/755141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/41/755141/1"}}},"commit":{"parents":[{"commit":"a19a2e824ad811b05ad7e4ffeab1ffbddc85f472","subject":"Merge \"Adding Erno Kuvaja candidacy for TC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a19a2e824ad811b05ad7e4ffeab1ffbddc85f472"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-09-30 00:08:46.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-09-30 00:08:46.000000000","tz":0},"subject":"Add a combined_end_nominations E-mail template","message":"Add a combined_end_nominations E-mail template\n\nSince the last time we ran a combined election there were no runoffs\nneeded, nobody had yet created an end_nominations template for the\ncombined model. This one is a mashup of the existing PTL and TC\ntemplates, with some patterns borrowed from combined templates for\nother events. Also included is a typo fix for the\ncombined_nominations_last_days template which I just noticed when\nreferencing it.\n\nChange-Id: Icfbff4ec284f227d2c575f43cf97e27190bca62e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8536c504ef7c443921411e8dcf102dbc8381a871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8536c504ef7c443921411e8dcf102dbc8381a871"}]},"branch":"refs/heads/master"},"119165be672d57aad4bbbadff8b3e6329feb491a":{"kind":"REWORK","_number":2,"created":"2020-10-26 14:46:26.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/41/755141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/41/755141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/41/755141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/41/755141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/41/755141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/41/755141/2"}}},"commit":{"parents":[{"commit":"a19a2e824ad811b05ad7e4ffeab1ffbddc85f472","subject":"Merge \"Adding Erno Kuvaja candidacy for TC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a19a2e824ad811b05ad7e4ffeab1ffbddc85f472"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-09-30 00:08:46.000000000","tz":0},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-10-26 14:46:18.000000000","tz":-420},"subject":"Add a combined_end_nominations E-mail template","message":"Add a combined_end_nominations E-mail template\n\nSince the last time we ran a combined election there were no runoffs\nneeded, nobody had yet created an end_nominations template for the\ncombined model. This one is a mashup of the existing PTL and TC\ntemplates, with some patterns borrowed from combined templates for\nother events. Also included is a typo fix for the\ncombined_nominations_last_days template which I just noticed when\nreferencing it.\n\nChange-Id: Icfbff4ec284f227d2c575f43cf97e27190bca62e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/119165be672d57aad4bbbadff8b3e6329feb491a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/119165be672d57aad4bbbadff8b3e6329feb491a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
