)]}'
{"id":"openstack%2Felection~761486","triplet_id":"openstack%2Felection~master~Id9eb333579d988a368bdf99ce6a01ac90bd3b1c0","project":"openstack/election","branch":"master","hashtags":[],"change_id":"Id9eb333579d988a368bdf99ce6a01ac90bd3b1c0","subject":"bump py37 to py38 in tox.ini in \u0027victoria\u0027 cycle, we should test py38 by default.","status":"ABANDONED","created":"2020-11-05 02:22:28.000000000","updated":"2021-07-23 16:48:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7438c2ef32365ffa52180dd27bc625f07b772646","_number":761486,"virtual_id_number":761486,"owner":{"_account_id":32211,"name":"kuang congxian","email":"kuangcx@inspur.com","username":"kuangcx"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-05 02:30:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-05 12:27:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 02:30:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-05 12:27:31.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"c8356b8a183aab8f9bd9f2e01fb1a1d067bb4215","author":{"_account_id":32211,"name":"kuang congxian","email":"kuangcx@inspur.com","username":"kuangcx"},"date":"2020-11-05 02:22:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2d9ca03177017c10baca1418e0d30d59caf0cab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 02:30:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07908ee6f01493bbeeb03a6d973d2fa : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a19226c872b4f14ac7d1951c06ef881 : SUCCESS in 2m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a528da609f29454d9dcdea8cea9701b0 : SUCCESS in 2m 45s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/4ca17ac32d534ee79b9ef3f1b4d323d5 : SUCCESS in 2m 41s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1996b88317d643bfa07c59eef1ea7ead : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fc984cc6ed007ad31aca77822b20dc1193a7139f","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-11-05 12:27:31.000000000","message":"Patch Set 1: Code-Review+1\n\nThis looks like a good change.","accounts_in_message":[],"_revision_number":1},{"id":"7438c2ef32365ffa52180dd27bc625f07b772646","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-07-23 16:48:41.000000000","message":"Abandoned\n\n(no action more than six months - author: please re-initiate this patch by setting with appropriate \"topic\" label would be helpful to move forward in future)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e34bc4528733475c418fc8fa7ee345bd10f68466","revisions":{"e34bc4528733475c418fc8fa7ee345bd10f68466":{"kind":"REWORK","_number":1,"created":"2020-11-05 02:22:28.000000000","uploader":{"_account_id":32211,"name":"kuang congxian","email":"kuangcx@inspur.com","username":"kuangcx"},"ref":"refs/changes/86/761486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/86/761486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/86/761486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/86/761486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/86/761486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/86/761486/1"}}},"commit":{"parents":[{"commit":"dd4725cccd0ef81e0eb24c25f1fcc9008117dfbd","subject":"Close out PTL and TC elections for Wallaby Cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dd4725cccd0ef81e0eb24c25f1fcc9008117dfbd"}]}],"author":{"name":"kuangcx","email":"kuangcx@inspur.com","date":"2020-11-05 02:21:36.000000000","tz":480},"committer":{"name":"kuangcx","email":"kuangcx@inspur.com","date":"2020-11-05 02:21:36.000000000","tz":480},"subject":"bump py37 to py38 in tox.ini in \u0027victoria\u0027 cycle, we should test py38 by default.","message":"bump py37 to py38 in tox.ini\nin \u0027victoria\u0027 cycle, we should test py38 by default.\n\nChange-Id: Id9eb333579d988a368bdf99ce6a01ac90bd3b1c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e34bc4528733475c418fc8fa7ee345bd10f68466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e34bc4528733475c418fc8fa7ee345bd10f68466"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
