)]}'
{"id":"openstack%2Felection~771077","triplet_id":"openstack%2Felection~master~Ifb37a00ca800e99699de2cf68e12533a4545f750","project":"openstack/election","branch":"master","topic":"setup-xena-election","hashtags":[],"change_id":"Ifb37a00ca800e99699de2cf68e12533a4545f750","subject":"Setup X Election","status":"MERGED","created":"2021-01-15 20:26:54.000000000","updated":"2021-03-02 13:20:28.000000000","submitted":"2021-03-02 04:33:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"771077-setup-xena-election","meta_rev_id":"b23b17e8a05015cadcd0ced5fea85f10ede60b65","_number":771077,"virtual_id_number":771077,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 04:33:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2021-03-02 04:29:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-02 13:20:28.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2021-03-02 04:05:56.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2021-03-02 04:29:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 20:33:00.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2021-01-15 20:39:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-15 20:40:53.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-01-15 20:52:40.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-01-18 19:16:09.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-02-26 05:49:16.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2021-02-26 18:41:20.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"CC"},{"updated":"2021-02-26 18:49:21.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"}],"messages":[{"id":"fb9bf5ee7dc9b247de1e9de0375fe70a5fd1853c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-01-15 20:26:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4956954228faa31c281cd5ca0eef873ee3f6767a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-15 20:33:00.000000000","message":"Patch Set 1:\n\nThis is probably also a good time to start thinking about messaging around necessary changes due to permissions in new Gerrit. When we query an account, we no longer get a list of all secondary E-mail addresses. As a result, it won\u0027t be sufficient for *some* E-mail address in Gerrit to correspond to an E-mail address in a foundation membership. Instead, folks who want a ballot will need to make sure their *preferred* E-mail address in Gerrit is included with their foundation member profile (doesn\u0027t have to be the primary foundation profile address, but it does need to be added to the profile).\n\nThe situation for candidates doesn\u0027t really change, I don\u0027t think, they still need the address associated with their candidacy to be associated with both their Gerrit account and their foundation membership, doesn\u0027t have to be preferred/primary addresses as we look them up by the supplied address.","accounts_in_message":[],"_revision_number":1},{"id":"401276da583023579d3b1c62aca7b210f93af84a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:39:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6414d8b502d4141b77c2314d488ae90 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b17f0d3bc464261be1e4a0b0943c632 : SUCCESS in 3m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/13c063bc0fe34ed09e8fc49fa337d976 : SUCCESS in 4m 06s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/6784d03867b246028dce0f6ba8019062 : SUCCESS in 3m 41s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c186e4ce38f4418e8fd10aa30a7fdcc9 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1e2d6902213eae142e5f73b7e8fa2555746f7cb0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-15 20:40:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aeade1227421d910ab8ca3f9faacb35d8429052c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-01-15 20:46:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b1fb4cebb3667dcdc53aadb1ccc848c40f894820","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-01-15 20:52:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nI think this is ok.  One question in-line for my education.","accounts_in_message":[],"_revision_number":2},{"id":"a44eb4d041732b41f1f0c90ae9631418f86d5bd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 20:59:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd43ba95814041599da6a1a17ac20e2a : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73a7b3c829834900937f331c0dbfba3a : SUCCESS in 3m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6c26944b2149463ba23def66721a72a3 : SUCCESS in 3m 09s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/526c715d1e8542a19f18a1662fa4d3ea : SUCCESS in 3m 21s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/211182afb97c4689bd421c1c1a7286cd : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"8717e3968b8e427baa5d46a2f70a679832ca292f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-18 19:16:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"de3531103c3713b9c592449b5ec0f6ba86433c1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-02-25 22:12:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6b3ff67e6366afaa6a1210aabea5f947e5687fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 22:21:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e15c420b834b3a91f5c04aae7509c1 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5497e7c1b0104075af4af4d1297134df : SUCCESS in 4m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/73114425419a4e73b17d8954e20d6077 : SUCCESS in 4m 53s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/947b39338d94472da3fb7010742750bc : SUCCESS in 3m 23s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2a9ea96879a242ea974e81967e77c525 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":3},{"id":"5f8e675cce7d1fcc0cf13c2831f7d67e2a6b056c","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-02-26 05:49:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"84afa503c73bd25272ee577c4f2a2401c198934a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-26 15:08:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82b870ab90848add6e6ebcbf95c4ba40808a3632","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-26 15:13:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"07f03c537725981ad28827066937888f1f1fe39b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-26 15:18:22.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2c6ee0804907d9013046be1f874defe61c15153a","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-02-26 18:39:49.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0b0d73f5d7be5f341b8e29ed5f64a1cf67a8e489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-02-26 18:41:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7ae7405ce9e5ddfa922b6f04263fc9a88ee46db2","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-02-26 18:41:20.000000000","message":"Assignee added: Mohammed Naser \u003cmnaser@vexxhost.com\u003e","accounts_in_message":[],"_revision_number":4},{"id":"a62e703996a8a871687d83fd32b55603de42f6e1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-26 18:45:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cf504c80951b3e78803e7dcc33789ffb292e9131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 18:47:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810c12a5234647b691a1d16acd5359fe : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8363a12aec0548baaeb530aea2157a5b : SUCCESS in 4m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/570a02ddb66149f3932040205d95e579 : SUCCESS in 4m 30s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/eef846dae6ee4d86a970eb150a2ac1b0 : SUCCESS in 3m 52s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/48e0afa5fa7c4905a0598bfb6c393019 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":4},{"id":"bb43afb35f6cba2cf58e40306f0d238f6c427b83","tag":"autogenerated:gerrit:deleteAssignee","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-02-26 18:49:08.000000000","message":"Assignee deleted: Mohammed Naser \u003cmnaser@vexxhost.com\u003e","accounts_in_message":[],"_revision_number":4},{"id":"8c2458110eba5b83cd51d0cba128249975613ae2","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-02-26 21:02:05.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\n(Raising a minor tick)","accounts_in_message":[],"_revision_number":4},{"id":"30475a62c0b29d902624b83c184caa40728f93a3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-26 21:08:16.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nLet\u0027s separate the Naser addition to election official or at least mention in commit msg.","accounts_in_message":[],"_revision_number":4},{"id":"8004c48b249ba57f24b8f4c6f3bdfe4fc805cb3c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-26 21:08:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"790c6960ef1dd2d659f4c02820b00cf756abd4eb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-26 21:09:15.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"a76862f1032722575d2d5b4b917981cb1be85609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-02-27 05:58:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0cdf499bb5062b7b777783be0e61babd2dd20487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 06:19:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb5b993e5644ae49bfeb4efb842f105 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3876c4ccffe4b4fb4bbbccf184c07a4 : SUCCESS in 3m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ec23f9d0eca14097a469e672eebf072a : SUCCESS in 3m 13s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/a35adb87fdf64688b2485756fb0c6211 : SUCCESS in 5m 54s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b3756eb2cc394e43a05ee34b23d26daa : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":5},{"id":"376d871c3e3a36d9b746bf5dd48c1e6771d3e5dc","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-27 14:06:24.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7429ce532fb59a7a1fed05a8b7553b72200644df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-03-01 23:19:10.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"10028c2b1db4bf4ee62430500162d902e146775f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 23:27:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a04676513394ad0bf4b6003687738e4 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d346db1c833a46cfafbb4c5305536216 : SUCCESS in 3m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4b7b35b9a9c94109883a637f20b8fd4c : SUCCESS in 4m 01s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/20be51d775da4ba78b0cd5938a0ab0c5 : SUCCESS in 3m 35s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/19808e3cbad447c98395d4802ceff792 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":6},{"id":"14f29e5b8a752c4acb0e0886fad860f8c5284725","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-01 23:53:57.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2227e1b10a2d7d68dcf50742266670c8353bf045","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-03-02 03:17:37.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"3045fedf77889aed3c4a68c184c82727e47718bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 03:26:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d741e92b8c64ed5b93fb4cc4c3c3f0d : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc3b32b974e84c8f8cd6bc127219cad4 : SUCCESS in 3m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8e419fa89d4644b98d13c5f8769b780a : SUCCESS in 3m 21s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/1afc134cd62944dc810a090a52ffac31 : SUCCESS in 3m 17s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a533c1ca593a4fad81620f17aa20cf40 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":7},{"id":"641641bdb6bb573743484bef7e4e0b3a686a8f40","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-02 04:05:56.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"6bd7b2283344d418c1533cf83464e630239f2915","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-03-02 04:29:10.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nGreat thank you","accounts_in_message":[],"_revision_number":7},{"id":"ee8175410aad3cbdb60754a5c3e15974204a6f9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 04:29:24.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d9752c30a90d21486e1a206743b6c2c2bcd775f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 04:33:54.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487b402262c9438ebfd3376864359fc0 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80ae4e257768462485d4fa3ec963ea70 : SUCCESS in 3m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/50226b00321d4703a29e8a6571839afe : SUCCESS in 3m 23s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/77b1fe8a8ec54a2bbc923893ddbda03b : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":7},{"id":"5f7618ac0de5061e4bb43d109e6caa44b92c4e76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 04:33:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"07a83598777aaec5c511c32d4c6ccceed935c490","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 04:35:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/0162b3331b3d4803ba70f65788ce104b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":7},{"id":"b23b17e8a05015cadcd0ced5fea85f10ede60b65","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-02 13:20:28.000000000","message":"Patch Set 7: Code-Review+1\n\nthanks, forgot to vote previously but dates lgtm now.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2b614dada6b0f622088fbb169776630db8b065dc","revisions":{"5fb6240bb22bce6e517f032305a21ecac9d92bcb":{"kind":"REWORK","_number":1,"created":"2021-01-15 20:26:54.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/77/771077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/77/771077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/77/771077/1"}}},"commit":{"parents":[{"commit":"761ca1c565b548e1a355464c7610c4a0809c9606","subject":"Remove Tony from list of election officials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/761ca1c565b548e1a355464c7610c4a0809c9606"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-01-15 20:23:15.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-01-15 20:26:47.000000000","tz":-480},"subject":"Setup X Election","message":"Setup X Election\n\nHonestly, running the elections in parallel is so much easier than running\nthem separately. Since we have the precendence for running them together, I\nfigured I would propose we do it again. Happy to change it though. I used\ntonyb\u0027s tool, but the dates for the begining of victoria and end of wallaby\nweren\u0027t right? Perhaps my arguments were wrong, but I think I have them\nright.\n\nChange-Id: Ifb37a00ca800e99699de2cf68e12533a4545f750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5fb6240bb22bce6e517f032305a21ecac9d92bcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5fb6240bb22bce6e517f032305a21ecac9d92bcb"}]},"branch":"refs/heads/master"},"91516d7452a19a50d4ad467467ad620b1a2794d1":{"kind":"REWORK","_number":2,"created":"2021-01-15 20:46:40.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/77/771077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/77/771077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/77/771077/2"}}},"commit":{"parents":[{"commit":"761ca1c565b548e1a355464c7610c4a0809c9606","subject":"Remove Tony from list of election officials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/761ca1c565b548e1a355464c7610c4a0809c9606"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-01-15 20:23:15.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-01-15 20:46:20.000000000","tz":0},"subject":"Setup X Election","message":"Setup X Election\n\nHonestly, running the elections in parallel is so much easier than running\nthem separately. Since we have the precendence for running them together, I\nfigured I would propose we do it again. Happy to change it though. I used\ntonyb\u0027s tool, but the dates for the begining of victoria and end of wallaby\nweren\u0027t right? Perhaps my arguments were wrong, but I think I have them\nright.\n\nChange-Id: Ifb37a00ca800e99699de2cf68e12533a4545f750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/91516d7452a19a50d4ad467467ad620b1a2794d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/91516d7452a19a50d4ad467467ad620b1a2794d1"}]},"branch":"refs/heads/master"},"2635549dcbdbf73136cd7a1c058621624b7e785c":{"kind":"REWORK","_number":3,"created":"2021-02-25 22:12:55.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/77/771077/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/77/771077/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/77/771077/3"}}},"commit":{"parents":[{"commit":"85151146095fd838079dc89dbdc8985b7ca21a7b","subject":"Merge \"Add Wallaby election stats\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/85151146095fd838079dc89dbdc8985b7ca21a7b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-01-15 20:23:15.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-25 22:12:49.000000000","tz":-480},"subject":"Setup X Election","message":"Setup X Election\n\nHonestly, running the elections in parallel is so much easier than running\nthem separately. Since we have the precendence for running them together, I\nfigured I would propose we do it again. Happy to change it though. I used\ntonyb\u0027s tool, but the dates for the begining of victoria and end of wallaby\nweren\u0027t right? Perhaps my arguments were wrong, but I think I have them\nright.\n\nChange-Id: Ifb37a00ca800e99699de2cf68e12533a4545f750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2635549dcbdbf73136cd7a1c058621624b7e785c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2635549dcbdbf73136cd7a1c058621624b7e785c"}]},"branch":"refs/heads/master"},"6f7acc855b873666b1d094e9aeeef910bd2309b8":{"kind":"REWORK","_number":4,"created":"2021-02-26 18:41:08.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/77/771077/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/77/771077/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/77/771077/4"}}},"commit":{"parents":[{"commit":"85151146095fd838079dc89dbdc8985b7ca21a7b","subject":"Merge \"Add Wallaby election stats\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/85151146095fd838079dc89dbdc8985b7ca21a7b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-01-15 20:23:15.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-26 18:41:02.000000000","tz":-480},"subject":"Setup X Election","message":"Setup X Election\n\nHonestly, running the elections in parallel is so much easier than running\nthem separately. Since we have the precendence for running them together, I\nfigured I would propose we do it again. Happy to change it though. I used\ntonyb\u0027s tool, but the dates for the begining of victoria and end of wallaby\nweren\u0027t right? Perhaps my arguments were wrong, but I think I have them\nright.\n\nChange-Id: Ifb37a00ca800e99699de2cf68e12533a4545f750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6f7acc855b873666b1d094e9aeeef910bd2309b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6f7acc855b873666b1d094e9aeeef910bd2309b8"}]},"branch":"refs/heads/master"},"e3ea788ceba5bfe9fab213c0daa494c1c6b3ecaf":{"kind":"REWORK","_number":5,"created":"2021-02-27 05:58:34.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/77/771077/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/77/771077/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/77/771077/5"}}},"commit":{"parents":[{"commit":"85151146095fd838079dc89dbdc8985b7ca21a7b","subject":"Merge \"Add Wallaby election stats\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/85151146095fd838079dc89dbdc8985b7ca21a7b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-01-15 20:23:15.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-27 05:57:57.000000000","tz":-480},"subject":"Setup X Election","message":"Setup X Election\n\nHonestly, running the elections in parallel is so much easier than running\nthem separately. Since we have the precendence for running them together, I\nfigured I would propose we do it again.\n\nChange-Id: Ifb37a00ca800e99699de2cf68e12533a4545f750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e3ea788ceba5bfe9fab213c0daa494c1c6b3ecaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e3ea788ceba5bfe9fab213c0daa494c1c6b3ecaf"}]},"branch":"refs/heads/master"},"981bd83f7292b021ed64aaeaafa6deb6a595b1d5":{"kind":"REWORK","_number":6,"created":"2021-03-01 23:19:10.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/77/771077/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/77/771077/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/77/771077/6"}}},"commit":{"parents":[{"commit":"85151146095fd838079dc89dbdc8985b7ca21a7b","subject":"Merge \"Add Wallaby election stats\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/85151146095fd838079dc89dbdc8985b7ca21a7b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-01-15 20:23:15.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-03-01 23:18:57.000000000","tz":0},"subject":"Setup X Election","message":"Setup X Election\n\nHonestly, running the elections in parallel is so much easier than running\nthem separately. Since we have the precendence for running them together, I\nfigured I would propose we do it again.\n\nChange-Id: Ifb37a00ca800e99699de2cf68e12533a4545f750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/981bd83f7292b021ed64aaeaafa6deb6a595b1d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/981bd83f7292b021ed64aaeaafa6deb6a595b1d5"}]},"branch":"refs/heads/master"},"2b614dada6b0f622088fbb169776630db8b065dc":{"kind":"REWORK","_number":7,"created":"2021-03-02 03:17:37.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/77/771077/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/77/771077/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/77/771077/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/77/771077/7"}}},"commit":{"parents":[{"commit":"85151146095fd838079dc89dbdc8985b7ca21a7b","subject":"Merge \"Add Wallaby election stats\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/85151146095fd838079dc89dbdc8985b7ca21a7b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-01-15 20:23:15.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-03-02 03:17:29.000000000","tz":0},"subject":"Setup X Election","message":"Setup X Election\n\nHonestly, running the elections in parallel is so much easier than running\nthem separately. Since we have the precendence for running them together, I\nfigured I would propose we do it again.\n\nChange-Id: Ifb37a00ca800e99699de2cf68e12533a4545f750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2b614dada6b0f622088fbb169776630db8b065dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2b614dada6b0f622088fbb169776630db8b065dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
