)]}'
{"id":"openstack%2Felection~777810","triplet_id":"openstack%2Felection~master~I93662e0f9904c78c19fb355e2ad47e53a0dd3bd8","project":"openstack/election","branch":"master","topic":"setup-xena-election","hashtags":[],"change_id":"I93662e0f9904c78c19fb355e2ad47e53a0dd3bd8","subject":"Setup Xena Directories","status":"MERGED","created":"2021-02-26 18:47:18.000000000","updated":"2021-03-02 18:58:45.000000000","submitted":"2021-03-02 18:57:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777810-setup-xena-election","meta_rev_id":"2b49add19a5519db4da89efb17f1b8034cad3894","_number":777810,"virtual_id_number":777810,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 18:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2021-03-02 18:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-02 18:36:36.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2021-03-02 18:50:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-26 19:08:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-26 21:22:24.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2021-03-02 13:40:12.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2021-03-02 16:35:53.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2021-03-02 18:36:36.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"85f7f99dd712147d96fd640fbf994460bf4bbcee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-02-26 18:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7846cb3561a20b55235cee95530be77b7f52f89c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 19:08:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8c9766d535847dfa4dd9068dbee156b : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9d2269dd3924232bc7d60352defae56 : SUCCESS in 3m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/174ee573d3b04b7b9a6738a0e7c7f057 : SUCCESS in 4m 17s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2ed2746a10254a2ea68a4ca37bcfaa56 : SUCCESS in 4m 39s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b061eceb11764cc895ecc42f333a69f1 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a911608d951b9b9cbef69c84405ca0a25f386239","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-26 21:22:24.000000000","message":"Patch Set 1:\n\nmay be we can remove the oslo dir from here as oslo follow DPL model and does not participate in PTL election.","accounts_in_message":[],"_revision_number":1},{"id":"c2da70c215ac7ee6e03a5378da80b372092b8831","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-03-02 04:50:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8492bb5c37570cfd022183e33776b597f1e88140","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 04:57:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aed74ea201a140a5a1c948fd0ee52467 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c4ed3bf36494f29a7f3bb9548331afd : SUCCESS in 4m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bebb343d5dc74e58810ef115abd9e07c : SUCCESS in 4m 49s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/6b3aa71d986d416ba92a4867809e64be : SUCCESS in 3m 52s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/bf0fffc10a9b45b09240f14468cb5c6a : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e903735b3e11068dd3ed05e259f2636de7e08800","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-02 13:40:12.000000000","message":"Patch Set 1:\n\nGhanshyam wrote:\n\n\u003e may be we can remove the oslo dir from here as oslo follow DPL model\n\u003e and does not participate in PTL election.\n\nI concur, https://governance.openstack.org/tc/reference/projects/oslo.html shows distributed leadership now. In the future, the election scripts could be made to detect that automatically.","accounts_in_message":[],"_revision_number":1},{"id":"5475ae41dff16fb90a0e8fdcbec13b7976cf253c","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-03-02 16:35:53.000000000","message":"Patch Set 2: Code-Review+2\n\nAgreed assuming that there is no currently auto-generated directories","accounts_in_message":[],"_revision_number":2},{"id":"e8a7bba19d112b29d21e956c231fae0ab4cfb450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-03-02 18:13:18.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"61c8444f638a25dcb19389e1f54afea9f60c10dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:23:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae6803c19b04e13b8de20382a5b71a1 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96d4f4b20332473bb715dad67b7050fb : SUCCESS in 4m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ef87bc0eaa6e4697bc8d02bb86e0af6f : SUCCESS in 6m 57s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/d22a5fa8fd294d668b79ad7bcc9e7fed : SUCCESS in 5m 45s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/33a49703ffce4305ae9d1f7397202743 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"f2b9f047f9ed3313b6163233723fecffd5f29996","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-02 18:36:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f0189b186ae5b35364f253e3962c7ab3c7f4bf03","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-03-02 18:50:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6c1d70764a0f785a8ca00561fea3aee805396eb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:50:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0895295703365f7ef289f4571f5d3246deea558b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:57:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f903b8b1d04c0bbc1503c6a1c13889 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e46c0a3378ea44ae9a6f09aa0113748f : SUCCESS in 4m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8c6bd61ad6a94ca58d3d298bea8ddc39 : SUCCESS in 3m 36s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c99eca04f77c4925b78186809694430f : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"8feb4359d1fad260aac515fc8d4759a9c86efdc7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:57:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2b49add19a5519db4da89efb17f1b8034cad3894","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:58:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/0b83edc5825a410896f2d309dd6d547b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fe70c27ca7e156efb521cd7845d5adb637b07c0e","revisions":{"9327db004e9655c558f64452606b5d4eb25e066e":{"kind":"REWORK","_number":1,"created":"2021-02-26 18:47:18.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/10/777810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/10/777810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/10/777810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/10/777810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/10/777810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/10/777810/1"}}},"commit":{"parents":[{"commit":"6f7acc855b873666b1d094e9aeeef910bd2309b8","subject":"Setup X Election","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6f7acc855b873666b1d094e9aeeef910bd2309b8"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-26 18:47:00.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-26 18:47:00.000000000","tz":-480},"subject":"Setup Xena Directories","message":"Setup Xena Directories\n\nChange-Id: I93662e0f9904c78c19fb355e2ad47e53a0dd3bd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9327db004e9655c558f64452606b5d4eb25e066e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9327db004e9655c558f64452606b5d4eb25e066e"}]},"branch":"refs/heads/master"},"1a3642cbb2a7caab6c50b9a4444522fba72a094d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-02 04:50:31.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/10/777810/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/10/777810/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/10/777810/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/10/777810/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/10/777810/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/10/777810/2"}}},"commit":{"parents":[{"commit":"2b614dada6b0f622088fbb169776630db8b065dc","subject":"Setup X Election","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2b614dada6b0f622088fbb169776630db8b065dc"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-26 18:47:00.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-03-02 04:50:31.000000000","tz":0},"subject":"Setup Xena Directories","message":"Setup Xena Directories\n\nChange-Id: I93662e0f9904c78c19fb355e2ad47e53a0dd3bd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1a3642cbb2a7caab6c50b9a4444522fba72a094d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1a3642cbb2a7caab6c50b9a4444522fba72a094d"}]},"branch":"refs/heads/master","description":"Rebase"},"fe70c27ca7e156efb521cd7845d5adb637b07c0e":{"kind":"REWORK","_number":3,"created":"2021-03-02 18:13:18.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/10/777810/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/10/777810/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/10/777810/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/10/777810/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/10/777810/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/10/777810/3"}}},"commit":{"parents":[{"commit":"2b614dada6b0f622088fbb169776630db8b065dc","subject":"Setup X Election","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2b614dada6b0f622088fbb169776630db8b065dc"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-02-26 18:47:00.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-03-02 18:13:15.000000000","tz":0},"subject":"Setup Xena Directories","message":"Setup Xena Directories\n\nChange-Id: I93662e0f9904c78c19fb355e2ad47e53a0dd3bd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fe70c27ca7e156efb521cd7845d5adb637b07c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fe70c27ca7e156efb521cd7845d5adb637b07c0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
