)]}'
{"id":"openstack%2Felection~779434","triplet_id":"openstack%2Felection~master~I2105a1e37f0611a56afa099995257a58271949dc","project":"openstack/election","branch":"master","hashtags":[],"change_id":"I2105a1e37f0611a56afa099995257a58271949dc","subject":"Add Chris MacNaughton candidacy for OpenStack Charms PTL","status":"MERGED","created":"2021-03-09 07:31:36.000000000","updated":"2021-03-09 18:12:51.000000000","submitted":"2021-03-09 18:12:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779434","meta_rev_id":"d413eec55f3a4c0a1b2f68d7ba7471601710b2d6","_number":779434,"virtual_id_number":779434,"owner":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 18:12:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2021-03-09 12:15:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2021-03-09 09:04:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-09 17:58:55.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-09 17:58:55.000000000","_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-09 07:38:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-09 09:04:31.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"state":"REVIEWER"},{"updated":"2021-03-09 12:15:14.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2021-03-09 17:58:55.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"6cbdcfc9a42d9221ba33d1086f07602cfb8e5539","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2021-03-09 07:31:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e051acc35c13ce7b1985815b8ae6ba4c0caef442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 07:38:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/713c6210433c4f2685d4ab13fa58f414 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1e14a07337d406faeff9985a0fa531e : SUCCESS in 3m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ca66f780b5bd48cca108564fece853c0 : SUCCESS in 5m 19s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/98698cb452954029b10ff8d7ff5bdff5 : SUCCESS in 3m 20s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/08239473e18b48b78f4610257f5f4502 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5c1f549a64344f485b19bc2852e1d92f73e9af85","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2021-03-09 09:04:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"464512d3212f24e2476ffc5c52979927df8ec7d4","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-03-09 12:15:14.000000000","message":"Patch Set 1: Code-Review+2\n\nValid candidate\n\n- OSF Member: https://openstack.org/community/members/profile/57059\n- Found: https://review.opendev.org/776332 merged on 2021-02-26 to openstack/charm-aodh for OpenStack Charms","accounts_in_message":[],"_revision_number":1},{"id":"f97d50d999785552af2139ab1850698173200bf9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-09 17:58:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI confirm this nominee\u0027s candidacy is valid.","accounts_in_message":[],"_revision_number":1},{"id":"074ce3af5a716c9a3b819b80ba560ddc6ebc7a95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 17:59:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df976ad283e3e68700431f241f140ca7ce9be504","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 18:12:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/596aec88c5ba4816a14d0c9ad3208e06 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cd06f1b38744d2d945acccb60062919 : SUCCESS in 7m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3fc4a287ee154b079830f0a88df40f1c : SUCCESS in 5m 03s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2c1d97ddf5d94ffc9d69a57505780e41 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d413eec55f3a4c0a1b2f68d7ba7471601710b2d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 18:12:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34b90782e65fbd67c712bad7fed29eb782cd2624","revisions":{"34b90782e65fbd67c712bad7fed29eb782cd2624":{"kind":"REWORK","_number":1,"created":"2021-03-09 07:31:36.000000000","uploader":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"ref":"refs/changes/34/779434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/34/779434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/34/779434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/34/779434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/34/779434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/34/779434/1"}}},"commit":{"parents":[{"commit":"ac7d5253151fe5e7cc930b16b71112b2975be3ba","subject":"Merge \"TC candidacy for mnaser\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ac7d5253151fe5e7cc930b16b71112b2975be3ba"}]}],"author":{"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","date":"2021-03-09 07:27:10.000000000","tz":60},"committer":{"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","date":"2021-03-09 07:32:18.000000000","tz":60},"subject":"Add Chris MacNaughton candidacy for OpenStack Charms PTL","message":"Add Chris MacNaughton candidacy for OpenStack Charms PTL\n\nChange-Id: I2105a1e37f0611a56afa099995257a58271949dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/34b90782e65fbd67c712bad7fed29eb782cd2624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/34b90782e65fbd67c712bad7fed29eb782cd2624"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
