)]}'
{"id":"openstack%2Felection~779911","triplet_id":"openstack%2Felection~master~I5e9231089bed76fafa6ded3915e2fed7f4bf7e41","project":"openstack/election","branch":"master","hashtags":[],"change_id":"I5e9231089bed76fafa6ded3915e2fed7f4bf7e41","subject":"Add Xinran Wang candidacy for Cyborg PTL in Xena","status":"ABANDONED","created":"2021-03-11 06:52:38.000000000","updated":"2021-03-26 00:12:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dd8e0a85d9b73b1c461ed1bc711e5e7ba9238468","_number":779911,"virtual_id_number":779911,"owner":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-11 07:51:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":1,"date":"2021-03-11 08:26:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2021-03-11 14:58:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 07:04:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 08:26:12.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2021-03-11 08:26:12.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-03-11 14:58:05.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"4ef5b43426847d6ed34bc8f51fb368f0d2e09d59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2021-03-11 06:52:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e615193a6d6ea9f4e2ddc3820f36e7abd1062a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 07:04:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf6a8deda02c47cc854d8a1adf4eb21b : FAILURE in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/483e0f0a189a435586b9a3cc143d9e04 : SUCCESS in 4m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/22e89da14b164c9b80195d33c5388bc7 : SUCCESS in 4m 25s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2af7819b97404c83b5cd6a442f7cc5dd : FAILURE in 3m 34s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/453d3054796d4988b73dc8e6b1ef50bf : FAILURE in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"dc9810d6e4b33c835bf2494d5b6e28a550cc19c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2021-03-11 07:26:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"46b78257de0646f7e6def0e9ddc8a3deabf3239c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 07:33:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bece0c0a40e4fd6aae554076236c3f6 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2064fffa6d8d453ca4cc0f5176680426 : SUCCESS in 4m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f55522684a7c4e9c8f7c7646a9489694 : SUCCESS in 4m 09s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/97e4fedd576d468aa2ddb8ddad173b80 : FAILURE in 4m 25s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/79aba44fbbd24f2392e8251130b0aaa2 : FAILURE in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"59e99e3d8b33faba1c47ae264eeeeea11d5354b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2021-03-11 07:43:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0c273b81e934d4f511b78cf84b09f60528715517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 07:51:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/692886d59b5e47bf922831026663f8ca : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d755b627577543f88509bcd857487a7a : SUCCESS in 4m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b4af1bf4397248b7958b851cc142f861 : SUCCESS in 4m 08s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/b6135d1297784f3193ca2698654707ef : SUCCESS in 4m 38s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5c5a58658a2942be800355540b77b53f : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":3},{"id":"1f035d41661c2b3e0614cfacd26e581038b0e9d9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-03-11 08:26:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"49ff65df51f7355a74a413ee418990623a5ca949","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-11 14:58:05.000000000","message":"Patch Set 3: Code-Review-2\n\nAs election officials we\u0027re unable to consider nominations arriving after the end of the scheduled nomination period (31 hours prior to your submission). At this point it\u0027s up to the TC to appoint PTLs for any teams which lack candidates, but we\u0027ll be sure to point out to them that you\u0027ve volunteered for the role.","accounts_in_message":[],"_revision_number":3},{"id":"c0801a6c4997754e863c1e12fce8de1ae86f0797","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-11 19:39:49.000000000","message":"Patch Set 3:\n\ncontinuing the TC discussion for late candidacy in https://review.opendev.org/c/openstack/governance/+/780071","accounts_in_message":[],"_revision_number":3},{"id":"dd8e0a85d9b73b1c461ed1bc711e5e7ba9238468","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-26 00:12:38.000000000","message":"Abandoned\n\nThe TC appointed you in change If778b0e878a00b9498a75a5ef74bd78574854ddc which has now merged.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c547af1682f338e278eb25cba22d5f419cf40c6e","revisions":{"147ff3d48548bf6ebda5b8b0c092edba35bd0b5d":{"kind":"REWORK","_number":1,"created":"2021-03-11 06:52:38.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/11/779911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/11/779911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/11/779911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/11/779911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/11/779911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/11/779911/1"}}},"commit":{"parents":[{"commit":"dc5046fe9d9563948d9db3b3b2a1efd2497713e4","subject":"Merge \"Add Lance Albertson candidancy for Openstack Chef PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dc5046fe9d9563948d9db3b3b2a1efd2497713e4"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2021-03-11 06:34:49.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2021-03-11 06:42:58.000000000","tz":0},"subject":"Add Xinran Wang candidacy for Cyborg PTL in Xena","message":"Add Xinran Wang candidacy for Cyborg PTL in Xena\n\nChange-Id: I5e9231089bed76fafa6ded3915e2fed7f4bf7e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/147ff3d48548bf6ebda5b8b0c092edba35bd0b5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/147ff3d48548bf6ebda5b8b0c092edba35bd0b5d"}]},"branch":"refs/heads/master"},"58eae282ad59aff2997277720caf139b7a01a82c":{"kind":"REWORK","_number":2,"created":"2021-03-11 07:26:27.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/11/779911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/11/779911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/11/779911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/11/779911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/11/779911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/11/779911/2"}}},"commit":{"parents":[{"commit":"dc5046fe9d9563948d9db3b3b2a1efd2497713e4","subject":"Merge \"Add Lance Albertson candidancy for Openstack Chef PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dc5046fe9d9563948d9db3b3b2a1efd2497713e4"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2021-03-11 06:34:49.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2021-03-11 07:16:31.000000000","tz":0},"subject":"Add Xinran Wang candidacy for Cyborg PTL in Xena","message":"Add Xinran Wang candidacy for Cyborg PTL in Xena\n\nChange-Id: I5e9231089bed76fafa6ded3915e2fed7f4bf7e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/58eae282ad59aff2997277720caf139b7a01a82c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/58eae282ad59aff2997277720caf139b7a01a82c"}]},"branch":"refs/heads/master"},"c547af1682f338e278eb25cba22d5f419cf40c6e":{"kind":"REWORK","_number":3,"created":"2021-03-11 07:43:17.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/11/779911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/11/779911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/11/779911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/11/779911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/11/779911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/11/779911/3"}}},"commit":{"parents":[{"commit":"dc5046fe9d9563948d9db3b3b2a1efd2497713e4","subject":"Merge \"Add Lance Albertson candidancy for Openstack Chef PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dc5046fe9d9563948d9db3b3b2a1efd2497713e4"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2021-03-11 06:34:49.000000000","tz":0},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2021-03-11 07:33:23.000000000","tz":0},"subject":"Add Xinran Wang candidacy for Cyborg PTL in Xena","message":"Add Xinran Wang candidacy for Cyborg PTL in Xena\n\nChange-Id: I5e9231089bed76fafa6ded3915e2fed7f4bf7e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c547af1682f338e278eb25cba22d5f419cf40c6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c547af1682f338e278eb25cba22d5f419cf40c6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
