)]}'
{"id":"openstack%2Felection~805332","triplet_id":"openstack%2Felection~master~I1eba298e9416ee8e2c4b5b6a5fb51ab0f0519ee9","project":"openstack/election","branch":"master","hashtags":[],"change_id":"I1eba298e9416ee8e2c4b5b6a5fb51ab0f0519ee9","subject":"Add Vishal Manchanda candidacy for Horizon PTL(Yoga)","status":"MERGED","created":"2021-08-20 05:22:45.000000000","updated":"2021-08-20 14:35:31.000000000","submitted":"2021-08-20 14:32:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805332","meta_rev_id":"2780ef82bb1ca5bda71cfb5540c7b09da3d15777","_number":805332,"virtual_id_number":805332,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-20 14:32:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2021-08-20 12:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-20 14:23:14.000000000","_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-20 05:32:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-20 12:37:21.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2021-08-20 14:23:14.000000000","updated_by":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"reviewer":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"state":"REVIEWER"}],"messages":[{"id":"2beb91639b3f334bcadd5ffb448f49a181df8587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-08-20 05:22:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e14c872d1f6828a8a7810c6e6dd79c8eb6c64356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 05:32:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/149280681ece4a66bde4305dc78e7fb3 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/227def78b4e94901b3a912321b0cfb7f : SUCCESS in 4m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/27aac68ba11449008fb50b7fbed2e57f : SUCCESS in 3m 38s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/257b2d58f18943819d54241b67fa9c41 : SUCCESS in 4m 29s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7500373aac1a472c9834c72264a196e9 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"86118cccd4dd367d805307a939f7075b5889b141","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-08-20 12:37:21.000000000","message":"Patch Set 1: Code-Review+2\n\nvalid PTL candidate - OSF member \u0026 https://review.opendev.org/779013 merged on 2021-08-02 to openstack/horizon for horizon","accounts_in_message":[],"_revision_number":1},{"id":"575c4dce2e7d77a80ca18686f8f5799be07c878e","author":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"},"date":"2021-08-20 14:23:14.000000000","message":"Patch Set 1: Workflow+1\n\nConfirmed the details Ian noted, and checked for additional merges:\nhttps://review.opendev.org/c/openstack/horizon/+/798730 - from Jul 28","accounts_in_message":[],"_revision_number":1},{"id":"0ae389a15f5d8179ce8c6b9854dd0a98eac2e926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:23:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b72663b8256f461bcfac2da9e18a8fd6a0373e99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:32:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa14f8632b3948788e74f893d82b1ae3 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d0dd9a07fd342cb9a2b27e5e344e4dd : SUCCESS in 4m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5bdd1ff33c3e44b5b543e0ce0cc25fd2 : SUCCESS in 4m 55s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7e025a79dbc94708aae289f2b8554898 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a40bc14005d4a24eceec35318f054826a84784ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:32:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2780ef82bb1ca5bda71cfb5540c7b09da3d15777","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:35:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/4b38d300c1f14feeb1b5a27a7adb1b2d : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2433fc02f3dc5aa9c1c2ae18666a273f301238b9","revisions":{"2433fc02f3dc5aa9c1c2ae18666a273f301238b9":{"kind":"REWORK","_number":1,"created":"2021-08-20 05:22:45.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/32/805332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/32/805332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/32/805332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/32/805332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/32/805332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/32/805332/1"}}},"commit":{"parents":[{"commit":"cd97623d06eabfff5d4504f13f71d5cee8d6ad1e","subject":"Merge \"Wang Hao candidacy for Zaqar PTL (Yoga)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/cd97623d06eabfff5d4504f13f71d5cee8d6ad1e"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-08-20 05:21:21.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-08-20 05:22:42.000000000","tz":330},"subject":"Add Vishal Manchanda candidacy for Horizon PTL(Yoga)","message":"Add Vishal Manchanda candidacy for Horizon PTL(Yoga)\n\nChange-Id: I1eba298e9416ee8e2c4b5b6a5fb51ab0f0519ee9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2433fc02f3dc5aa9c1c2ae18666a273f301238b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2433fc02f3dc5aa9c1c2ae18666a273f301238b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":2799,"name":"Andy McCrae","email":"andy.mccrae@googlemail.com","username":"andrew-mccrae"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
