)]}'
{"id":"openstack%2Felection~805512","triplet_id":"openstack%2Felection~master~I55138ccd685218268c1d126c3b2dee536faa2f5b","project":"openstack/election","branch":"master","hashtags":[],"change_id":"I55138ccd685218268c1d126c3b2dee536faa2f5b","subject":"Add Rong Zhu\u0027s candidacy for Solum PTL (Yoga)","status":"MERGED","created":"2021-08-22 03:02:36.000000000","updated":"2021-08-23 15:46:52.000000000","submitted":"2021-08-23 15:46:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805512","meta_rev_id":"9cf52b5a4194a5eb1ad88bc91db39bfde730a9f6","_number":805512,"virtual_id_number":805512,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-23 15:46:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2021-08-22 06:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2021-08-23 15:36:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2021-08-23 15:36:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-22 03:16:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-22 06:06:07.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2021-08-23 15:36:58.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"}],"messages":[{"id":"2233b677ae13c5e97a0bd226fb8a0ffb1dc2f245","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-08-22 03:02:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"858546ffb05fe55161cd86c60c848819c8ce20d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-08-22 03:04:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0ad5c62ae378f5d122e045c0fb96ab75836f056d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-08-22 03:04:31.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"7d7bda20ce2332f63995a4006634f86346a06b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 03:16:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11bad0ca49d7474ca2d9002a5fb5c3c5 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfd43b025171490fa81b8b11c7b34ee2 : SUCCESS in 3m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/27df25306c5d4616b395d62d6ca566a7 : SUCCESS in 3m 41s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/25970cd811da41d79a9da5b1cc365af9 : SUCCESS in 3m 37s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f448bb0d5bee4c3086abdef5bda19588 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":2},{"id":"4f8f725696798ecd8f6eb2961c798aa33ef34618","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-08-22 06:06:07.000000000","message":"Patch Set 2: Code-Review+2\n\nValid PTL candidate - OSF member: https://openstack.org/community/members/profile/28065 and https://review.opendev.org/784990 merged on 2021-04-07 to openstack/python-solumclient for solum","accounts_in_message":[],"_revision_number":2},{"id":"ed015e430e6e6bcf997a5937005ece8a8d2e824c","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2021-08-23 15:36:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ce424848ce6fd9224f6b1d92cd082f0e118457c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 15:37:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"32e9d7d9ad0593ca2398461ec6ef1171504915da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 15:46:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15c7cc5fb409443d891989454d73028c : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0f76d6d53a74f3f8d5ca75e7d6073d4 : SUCCESS in 5m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f31e4ac5f37242129a10e5fb07a42d6c : SUCCESS in 5m 18s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ec13a06924f74b07a52f067e6838529f : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"9cf52b5a4194a5eb1ad88bc91db39bfde730a9f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 15:46:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1fad79aaa7d1a81f5834c48ce91a5872d3a76e7d","revisions":{"9995645a09147f8f18d8cabdd7d69dded521c806":{"kind":"REWORK","_number":1,"created":"2021-08-22 03:02:36.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/12/805512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/12/805512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/12/805512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/12/805512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/12/805512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/12/805512/1"}}},"commit":{"parents":[{"commit":"516bb994c819f3ab9e64dc6095f2c80e662d8470","subject":"Merge \"Add Eyal Bar-Ilan candidacy for Vitrage PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/516bb994c819f3ab9e64dc6095f2c80e662d8470"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2021-08-22 03:02:36.000000000","tz":0},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2021-08-22 03:02:36.000000000","tz":0},"subject":"Add Rong Zhu\u0027s candidacy for Solum PTL (Yoga)","message":"Add Rong Zhu\u0027s candidacy for Solum PTL (Yoga)\n\nChange-Id: I55138ccd685218268c1d126c3b2dee536faa2f5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9995645a09147f8f18d8cabdd7d69dded521c806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9995645a09147f8f18d8cabdd7d69dded521c806"}]},"branch":"refs/heads/master"},"1fad79aaa7d1a81f5834c48ce91a5872d3a76e7d":{"kind":"REWORK","_number":2,"created":"2021-08-22 03:04:18.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/12/805512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/12/805512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/12/805512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/12/805512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/12/805512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/12/805512/2"}}},"commit":{"parents":[{"commit":"516bb994c819f3ab9e64dc6095f2c80e662d8470","subject":"Merge \"Add Eyal Bar-Ilan candidacy for Vitrage PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/516bb994c819f3ab9e64dc6095f2c80e662d8470"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2021-08-22 03:02:36.000000000","tz":0},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2021-08-22 03:04:05.000000000","tz":0},"subject":"Add Rong Zhu\u0027s candidacy for Solum PTL (Yoga)","message":"Add Rong Zhu\u0027s candidacy for Solum PTL (Yoga)\n\nChange-Id: I55138ccd685218268c1d126c3b2dee536faa2f5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1fad79aaa7d1a81f5834c48ce91a5872d3a76e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1fad79aaa7d1a81f5834c48ce91a5872d3a76e7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
