)]}'
{"id":"openstack%2Felection~805659","triplet_id":"openstack%2Felection~master~I23fa47835267a1d0fb2aba8b515a9ec7d5c5d797","project":"openstack/election","branch":"master","topic":"jsb-Yoga-TC-Nomination","hashtags":[],"change_id":"I23fa47835267a1d0fb2aba8b515a9ec7d5c5d797","subject":"Add Jay Bryant Yoga TC Candidacy","status":"MERGED","created":"2021-08-23 16:13:51.000000000","updated":"2021-08-23 16:40:41.000000000","submitted":"2021-08-23 16:39:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805659-jsb-Yoga-TC-Nomination","meta_rev_id":"8b1944a6da11fca2c85b2f37dd7c2a0cf3e5478e","_number":805659,"virtual_id_number":805659,"owner":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-23 16:39:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2021-08-23 16:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2021-08-23 16:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2021-08-23 16:26:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-23 16:15:36.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2021-08-23 16:23:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-23 16:26:34.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"6afc83b533d2698bb2c412e52f534f3e23eb0e9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-08-23 16:13:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b69ba034203aa1ad60934dd6b6633172532912f1","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2021-08-23 16:15:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7433cd818c48f847e3958143b82510c67732c46b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 16:23:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/834baabf382c48d0b78765f78efbd7cf : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce3d51e80c1448a691de54d138ff70bb : SUCCESS in 5m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/52f55639c0464880bfcff46f46859ae3 : SUCCESS in 5m 26s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/88a8858432e94a0ea21d5b1b92db4ae5 : SUCCESS in 4m 22s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/9c7ff114049f475eb946d715c5959890 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"abdfbf2bc6532bf2e362e4a022b7d7748c13d818","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-08-23 16:26:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nValid TC candidate: https://openstack.org/community/members/profile/8348","accounts_in_message":[],"_revision_number":1},{"id":"54b15a0cdc941ecc1f184da61f233315002d6026","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 16:26:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8b15acfa4b0e70e032d1f9b79aa437eaf5d2147","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 16:39:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac57219ea9054ea8a3427b8f95e20348 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ee3b485ba22414aad1c4fc6e5387fc2 : SUCCESS in 4m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6b3cd86f05cc4d1da795850bf4776e62 : SUCCESS in 4m 14s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/cbbc56e1a6f44c2aaf5d248a85cc883c : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"dca6d993cf6f3d7a4153123fa3290211f5c1c7c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 16:39:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8b1944a6da11fca2c85b2f37dd7c2a0cf3e5478e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 16:40:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/50afbf3469a84db3add6a7f60f5b1938 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7381f90682edad14eeee62c9593e2889160710b9","revisions":{"7381f90682edad14eeee62c9593e2889160710b9":{"kind":"REWORK","_number":1,"created":"2021-08-23 16:13:51.000000000","uploader":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"ref":"refs/changes/59/805659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/59/805659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/59/805659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/59/805659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/59/805659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/59/805659/1"}}},"commit":{"parents":[{"commit":"1d97d8cdeafc961becf851d194706997ba959384","subject":"Merge \"Add TC candidacy for Ghanshyam Mann\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1d97d8cdeafc961becf851d194706997ba959384"}]}],"author":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2021-08-23 16:12:41.000000000","tz":-300},"committer":{"name":"Jay S. Bryant","email":"jsbryant@electronicjungle.net","date":"2021-08-23 16:12:41.000000000","tz":-300},"subject":"Add Jay Bryant Yoga TC Candidacy","message":"Add Jay Bryant Yoga TC Candidacy\n\nI am announcing my Technical Committee\ncandidacy for the Yoga cycle.\n\nChange-Id: I23fa47835267a1d0fb2aba8b515a9ec7d5c5d797\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7381f90682edad14eeee62c9593e2889160710b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7381f90682edad14eeee62c9593e2889160710b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
