)]}'
{"id":"openstack%2Felection~805662","triplet_id":"openstack%2Felection~master~I4e696ed130da77c7d6edad8b16aca31e79310810","project":"openstack/election","branch":"master","hashtags":[],"change_id":"I4e696ed130da77c7d6edad8b16aca31e79310810","subject":"Adding Gregory Thiemonge candidacy for Octavia PTL","status":"MERGED","created":"2021-08-23 16:35:56.000000000","updated":"2021-08-23 17:06:18.000000000","submitted":"2021-08-23 17:02:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805662","meta_rev_id":"af31dc8be957b418243b2054e837d7cade00204c","_number":805662,"virtual_id_number":805662,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-23 17:02:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2021-08-23 16:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2021-08-23 16:36:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2021-08-23 16:55:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-23 16:36:44.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2021-08-23 16:53:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-23 16:55:03.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"a1509e685be598b0478497a88c375efbaf012d6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-23 16:35:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc5ccf03be6bc9642a2a1b4bb3656269ab4bc3b8","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2021-08-23 16:36:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1af91c3d4b94314193204c3bbda9ecb3e4dc4fdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 16:53:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241e7ef7aa8e49d3b3c20d035470e9db : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8466e47c0d3847ce9a910c06b3d5a972 : SUCCESS in 4m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3b7fe9d1826347f79baed29810dda6f5 : SUCCESS in 3m 54s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/db0b33dafb654d66bc33c9a57cad8aea : SUCCESS in 5m 21s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2604d051d8a84885a67f50cac6f237c6 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0781d2231fe903386b7c3e8880239328a22d46d8","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-08-23 16:55:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nValid candidate - https://openstack.org/community/members/profile/127095 + https://review.opendev.org/787528 merged on 2021-08-19 to openstack/octavia for octavia","accounts_in_message":[],"_revision_number":1},{"id":"9bb55b9ec6680796cd78b7715e2af4987472cf2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 16:55:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9698be4ef3568ae4938c3a09fba3c24af75b5bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 17:02:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4801042543ad4de3ac62ed4c3f895c5e : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bcff41c514b4490a12daf0b474dbd21 : SUCCESS in 5m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d7807f7cc72a45fda578e404a4734e8f : SUCCESS in 5m 05s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7442ee3b5643484a819ba259d4d4a30b : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ff5f844426c201ce989ca74c47ff2b268a3824cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 17:02:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"af31dc8be957b418243b2054e837d7cade00204c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 17:06:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/f8145a97db1e4d5b8766acc363860712 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd1c5b7652f6c1c7fe04b3c33e055a3f84458b0a","revisions":{"dd1c5b7652f6c1c7fe04b3c33e055a3f84458b0a":{"kind":"REWORK","_number":1,"created":"2021-08-23 16:35:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/62/805662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/62/805662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/62/805662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/62/805662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/62/805662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/62/805662/1"}}},"commit":{"parents":[{"commit":"cea0a97d0599393a542d1024112c7228192cdd45","subject":"Merge \"Add Yasufumi Ogawa candidacy for Tacker PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/cea0a97d0599393a542d1024112c7228192cdd45"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-23 16:32:33.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-08-23 16:32:45.000000000","tz":120},"subject":"Adding Gregory Thiemonge candidacy for Octavia PTL","message":"Adding Gregory Thiemonge candidacy for Octavia PTL\n\nChange-Id: I4e696ed130da77c7d6edad8b16aca31e79310810\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dd1c5b7652f6c1c7fe04b3c33e055a3f84458b0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dd1c5b7652f6c1c7fe04b3c33e055a3f84458b0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
