)]}'
{"id":"openstack%2Felection~805867","triplet_id":"openstack%2Felection~master~Ifb86d6550de6a4ff2894cb1a31c81103d482871c","project":"openstack/election","branch":"master","topic":"cinder-rosmaita-y","hashtags":[],"change_id":"Ifb86d6550de6a4ff2894cb1a31c81103d482871c","subject":"Adding Brian Rosmaita candidacy for Cinder PTL","status":"MERGED","created":"2021-08-24 15:21:22.000000000","updated":"2021-08-24 18:56:59.000000000","submitted":"2021-08-24 18:55:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805867-cinder-rosmaita-y","meta_rev_id":"d096da81cb88d4d516f133e442727c08002b4ba3","_number":805867,"virtual_id_number":805867,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 18:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2021-08-24 18:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2021-08-24 17:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2021-08-24 18:46:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 15:32:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 17:31:36.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2021-08-24 18:46:42.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"e2130627a78aae4996f73a57b21cf337b865d9c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-24 15:21:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa085f2274553017ce3dd0968c9f50f1c11e879f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 15:32:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18993de1cea04008a62e721d950afaca : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51a4ae76066245cb8b9e2c29e04979c2 : SUCCESS in 5m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8790f3039e714dc98d80cb5b3314e572 : SUCCESS in 5m 29s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/272a7d900185463c97d3cfbc46ec1d7d : SUCCESS in 5m 24s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/9b658e7b9a9f4e5fab43ab73934ec942 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c0eaeeedf3a02f703781b925ffbae8fd1ca6cb87","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2021-08-24 17:31:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f810ed5f6ff4d03bb4ed31888efbdb36e16baa6","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-08-24 18:46:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nhttps://openstack.org/community/members/profile/1778 \u0026  https://review.opendev.org/792642 merged on 2021-08-04 to openstack/cinder for cinder","accounts_in_message":[],"_revision_number":1},{"id":"8c89dd2c1dd50e3d42ff9a4d9c0ec9d8143a2471","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 18:47:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e166c1cc069fd486ec525baa3e4a044bc162f45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 18:55:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74fbcd4c2d949d3b07c4d98958527ff : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cab1e809d87a4c728c7d657bacd0e08c : SUCCESS in 4m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b8e93c7fe20445549bd99d503a7db898 : SUCCESS in 5m 33s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/709f2bb24b8b441481ed972937141ea7 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3649dbae385b7c5dbaf532a21203e7610cfaef6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 18:55:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d096da81cb88d4d516f133e442727c08002b4ba3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 18:56:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/8cf1ba48f6f34079a164d2ff6930d8f7 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb2e14ca3d2ab8832907c712bef78c967c41f048","revisions":{"fb2e14ca3d2ab8832907c712bef78c967c41f048":{"kind":"REWORK","_number":1,"created":"2021-08-24 15:21:22.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/67/805867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/67/805867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/67/805867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/67/805867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/67/805867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/67/805867/1"}}},"commit":{"parents":[{"commit":"165bd76615fac8116d6a03ec401ee620f812b314","subject":"Merge \"Add Iury Gregory candidacy for Ironic PTL (Yoga)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/165bd76615fac8116d6a03ec401ee620f812b314"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-08-24 14:11:49.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-08-24 14:43:42.000000000","tz":-240},"subject":"Adding Brian Rosmaita candidacy for Cinder PTL","message":"Adding Brian Rosmaita candidacy for Cinder PTL\n\nChange-Id: Ifb86d6550de6a4ff2894cb1a31c81103d482871c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fb2e14ca3d2ab8832907c712bef78c967c41f048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fb2e14ca3d2ab8832907c712bef78c967c41f048"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
