)]}'
{"id":"openstack%2Felection~805881","triplet_id":"openstack%2Felection~master~I5eb64c66fdecb5a21845573e7401272a7f2a9a70","project":"openstack/election","branch":"master","hashtags":[],"change_id":"I5eb64c66fdecb5a21845573e7401272a7f2a9a70","subject":"Adding Alex Kavanagh candidacy for OpenStack_Charms","status":"MERGED","created":"2021-08-24 19:26:31.000000000","updated":"2021-08-24 20:18:25.000000000","submitted":"2021-08-24 20:17:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805881","meta_rev_id":"58eee16704bc4172be9063849a429e7da9f6591a","_number":805881,"virtual_id_number":805881,"owner":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 20:17:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2021-08-24 20:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2021-08-24 19:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2021-08-24 20:10:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 19:33:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 19:45:15.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2021-08-24 20:10:42.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"1dd68b54b9e9f75a123884405d88227774c88417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-08-24 19:26:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ec4105041b8b1dac1747e548649244759e4b9e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 19:33:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e25c4ce6a09244dfb570ef6fb028894d : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/094762c729564a04876879e708494f92 : SUCCESS in 5m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2dee0547a7f340098bf0fdd6d9d2b53a : SUCCESS in 4m 30s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/759667d9d3304d39bc6040ef31a831a0 : SUCCESS in 4m 20s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/d0ff2dac9062495496fc19d6023287af : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"adb5a11dccfe0bbc33f100158749ed930f327ee4","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2021-08-24 19:45:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"99db2973650534d3a80fc22e70a04522a1b0b362","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-08-24 20:10:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nValid candidate - https://openstack.org/community/members/profile/57503 and https://review.opendev.org/792881 merged on 2021-06-28 to openstack/charms.ceph for OpenStack Charms","accounts_in_message":[],"_revision_number":1},{"id":"fa2456fbeb658a5c7c5c8a62a9a96573a3579f10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 20:11:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98f07629055c1b8a8ec2d5a98f18a3dddcc38d6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 20:17:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d64685677ee3493188f16c95e8f8fff1 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e743f5eaa2c4f4aa812b7ba42420ce1 : SUCCESS in 4m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/27e9af125abf41399b2afa6f26ef184d : SUCCESS in 4m 22s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/6bcfc51611e944eeb023444f5c236cd1 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1d4876e874f1c147cfad7133600f1a06206c9b30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 20:17:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"58eee16704bc4172be9063849a429e7da9f6591a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 20:18:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/871742d24fea4abab090d75a3a7cf4b0 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbbf96a0ef2f0d87623acc7971aca2a09223bef9","revisions":{"dbbf96a0ef2f0d87623acc7971aca2a09223bef9":{"kind":"REWORK","_number":1,"created":"2021-08-24 19:26:31.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/81/805881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/805881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/805881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/805881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/805881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/805881/1"}}},"commit":{"parents":[{"commit":"37500d2d7b08239653a117a459eeb2234e367cea","subject":"Merge \"Adding Takashi Kajinami candidacy for Storlets PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/37500d2d7b08239653a117a459eeb2234e367cea"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-08-24 19:26:14.000000000","tz":60},"committer":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-08-24 19:26:29.000000000","tz":60},"subject":"Adding Alex Kavanagh candidacy for OpenStack_Charms","message":"Adding Alex Kavanagh candidacy for OpenStack_Charms\n\nChange-Id: I5eb64c66fdecb5a21845573e7401272a7f2a9a70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dbbf96a0ef2f0d87623acc7971aca2a09223bef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dbbf96a0ef2f0d87623acc7971aca2a09223bef9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
