)]}'
{"id":"openstack%2Felection~805915","triplet_id":"openstack%2Felection~master~I01e41728375f2b6c9a241ec6b27ecb8429a6cd00","project":"openstack/election","branch":"master","hashtags":[],"change_id":"I01e41728375f2b6c9a241ec6b27ecb8429a6cd00","subject":"Tim Burke candidacy for Swift PTL (Yoga)","status":"MERGED","created":"2021-08-24 22:57:57.000000000","updated":"2021-08-25 00:01:46.000000000","submitted":"2021-08-24 23:59:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805915","meta_rev_id":"af4b595bf410dfd32941b6dc9d4665d09468bd8b","_number":805915,"virtual_id_number":805915,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 23:59:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2021-08-24 23:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2021-08-24 23:46:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 23:12:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 23:33:40.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"c938b67906fe9fa7bbb49b81bfd9c6d442bf8831","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-08-24 22:57:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79e6f301e5e8ad159e0ca230471cbc95c9df312e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 23:12:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b04c15ecbf7400aab6285991642ef61 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df1bb83fe36c491ea913a19d10af8512 : SUCCESS in 7m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1e630c1391534e9b9ed660b70754a494 : SUCCESS in 5m 28s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/ee2e8e2acc354dbaa13ade5589cf5cda : SUCCESS in 3m 59s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/682811644baa40af936dfc44eff76ab4 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c9c87af9478b49e0ca0ebb839b3c18753a80423d","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-08-24 23:33:40.000000000","message":"Patch Set 1: Code-Review+2\n\nValid - https://openstack.org/community/members/profile/31487 and  https://review.opendev.org/794137 merged on 2021-07-26 to openstack/liberasurecode for swift","accounts_in_message":[],"_revision_number":1},{"id":"d697c2eb81fadf137799f445006830ee27941f43","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-08-24 23:46:29.000000000","message":"Patch Set 1: Workflow+1\n\n(Single core +A considering current official availability + next stage needs to start very soon)","accounts_in_message":[],"_revision_number":1},{"id":"360d94c0feff6c85b4cb095579892de66f188659","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 23:46:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6004133b2d7d8aa5eadf852c1bccd33fb94843c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 23:59:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c583a7cc2ef9488b8a1dfa87277d5a4f : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53c3edd26f2948f4b9c2c06cdffcb853 : SUCCESS in 4m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/87e0804292064096903e41d815a8d67c : SUCCESS in 4m 22s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/14660ab3dc6b455e9f6bcc355fac6477 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"17bd6631704dd72e7c81dcc4ebd4fa289c52518d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 23:59:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"af4b595bf410dfd32941b6dc9d4665d09468bd8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 00:01:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/25942cef4a95453bbc8e65ce561e9bce : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9043c6c9ec3280194b71fd697b54a9e18391634e","revisions":{"9043c6c9ec3280194b71fd697b54a9e18391634e":{"kind":"REWORK","_number":1,"created":"2021-08-24 22:57:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/805915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/15/805915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/15/805915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/15/805915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/15/805915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/15/805915/1"}}},"commit":{"parents":[{"commit":"8585ce0e299542c023b075a5e432d52b07dfb204","subject":"Merge \"Adding Alex Kavanagh candidacy for OpenStack_Charms\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8585ce0e299542c023b075a5e432d52b07dfb204"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-08-24 22:57:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-08-24 22:57:33.000000000","tz":-420},"subject":"Tim Burke candidacy for Swift PTL (Yoga)","message":"Tim Burke candidacy for Swift PTL (Yoga)\n\nChange-Id: I01e41728375f2b6c9a241ec6b27ecb8429a6cd00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9043c6c9ec3280194b71fd697b54a9e18391634e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9043c6c9ec3280194b71fd697b54a9e18391634e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
