)]}'
{"id":"openstack%2Felection~806145","triplet_id":"openstack%2Felection~master~I2fe4585b281e26a90fec005f9c930b087d1d18a1","project":"openstack/election","branch":"master","hashtags":[],"change_id":"I2fe4585b281e26a90fec005f9c930b087d1d18a1","subject":"Adding Martin Chacon Piza candidacy for Monasca PTL for Y","status":"ABANDONED","created":"2021-08-26 11:49:02.000000000","updated":"2021-09-10 00:24:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f02c273f0c03ae214e3a7c10a1d6f99f520bcdf2","_number":806145,"virtual_id_number":806145,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-26 12:04:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":-2,"date":"2021-08-27 21:42:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 12:04:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-27 21:42:28.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2021-09-08 15:13:49.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"4d98f2766034196239e442d33eff61b84639f7a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-08-26 11:49:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1d73a2be698718e9858c6128ee99f589d7876f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:04:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255ad9b33d7c4e2287ddfadd6cc76360 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c12c31f81c949f8ad994bc07a0dd08e : SUCCESS in 5m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/11c8dd45e25a4758b092fa853657550c : SUCCESS in 5m 14s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/c7d54723e2dd44c6a1727f60d7c180a5 : SUCCESS in 5m 40s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/951ca988948549fd97d7e4f630016394 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6892b927499ffeaa3737aa92e3217dbcd04277d0","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-08-27 21:42:28.000000000","message":"Patch Set 1: Code-Review-2\n\nAs election officials we\u0027re unable to consider nominations arriving after the end of the scheduled nomination period (Aug 17, 2021 23:45 UTC - Aug 24, 2021 23:45 UTC). At this point it\u0027s up to the TC to appoint PTLs for any teams which lack candidates.","accounts_in_message":[],"_revision_number":1},{"id":"9ef4a1f974a1e35e82f2d40a9048562da782a17d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-08 15:13:49.000000000","message":"Patch Set 1:\n\nPTL appointment from TC : https://review.opendev.org/c/openstack/governance/+/807889/1","accounts_in_message":[],"_revision_number":1},{"id":"f02c273f0c03ae214e3a7c10a1d6f99f520bcdf2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-09-10 00:24:01.000000000","message":"Abandoned\n\nOwner \u0026 reviewers: please move discussions to https://review.opendev.org/c/openstack/governance/+/807889/ as TC proposed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1140898e54979d0f411ab3fa9e1bb84e0dc265d","revisions":{"e1140898e54979d0f411ab3fa9e1bb84e0dc265d":{"kind":"REWORK","_number":1,"created":"2021-08-26 11:49:02.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/45/806145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/45/806145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/45/806145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/45/806145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/45/806145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/45/806145/1"}}},"commit":{"parents":[{"commit":"d54468fa40402137e1909561a168b0bf099e5334","subject":"Merge \"bump py35,py37 to py3 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d54468fa40402137e1909561a168b0bf099e5334"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-08-26 11:47:51.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-08-26 11:47:51.000000000","tz":120},"subject":"Adding Martin Chacon Piza candidacy for Monasca PTL for Y","message":"Adding Martin Chacon Piza candidacy for Monasca PTL for Y\n\nChange-Id: I2fe4585b281e26a90fec005f9c930b087d1d18a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e1140898e54979d0f411ab3fa9e1bb84e0dc265d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e1140898e54979d0f411ab3fa9e1bb84e0dc265d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
