)]}'
{"id":"openstack%2Felection~823958","triplet_id":"openstack%2Felection~master~Ib3ce7b13041c83475ba2da28c6c9eacbf5e4b37b","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"33881":{"account":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"last_update":"2022-08-04 02:52:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib3ce7b13041c83475ba2da28c6c9eacbf5e4b37b","subject":"remove unicode from code","status":"ABANDONED","created":"2022-01-10 10:55:41.000000000","updated":"2022-08-04 02:52:45.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bfefce7b06466d04618939335b46c10ccb1dbbd4","_number":823958,"virtual_id_number":823958,"owner":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-10 11:02:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2022-01-27 16:34:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-10 11:02:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-27 16:34:02.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"87103df3c7f52b1120aa4e7ed83b6dfe71c307b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-01-10 10:55:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44291338f48d38698fd9bfb6ce7dc99bcc35edc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 11:02:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad5131c0285446689bb1abf315878b5 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51761ecfa18c4c3880c661ff40a4a831 : SUCCESS in 3m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7f85b61669af45b88bf3ff3e1c0f0d80 : SUCCESS in 3m 25s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/112e9f5cab71412689037513d37feb75 : SUCCESS in 3m 17s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/089aca4ac0c545ddb45ea5069599b226 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"fd39595dab8b219020d944a9cae84cdb1a0bff0b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-27 16:34:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfefce7b06466d04618939335b46c10ccb1dbbd4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-08-04 02:52:45.000000000","message":"Abandoned\n\nduplicate","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f9b7d0b48b82a12c3e2e8d5d17c014bcd44d61e","revisions":{"4f9b7d0b48b82a12c3e2e8d5d17c014bcd44d61e":{"kind":"REWORK","_number":1,"created":"2022-01-10 10:55:41.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/58/823958/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/58/823958/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/58/823958/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/58/823958/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/58/823958/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/58/823958/1"}}},"commit":{"parents":[{"commit":"704c0bf7ef0871abbcffed8f97e88787cce8fb52","subject":"corrected email","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/704c0bf7ef0871abbcffed8f97e88787cce8fb52"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-01-10 10:43:12.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-01-10 10:43:12.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: Ib3ce7b13041c83475ba2da28c6c9eacbf5e4b37b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4f9b7d0b48b82a12c3e2e8d5d17c014bcd44d61e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4f9b7d0b48b82a12c3e2e8d5d17c014bcd44d61e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
