)]}'
{"id":"openstack%2Felection~828279","triplet_id":"openstack%2Felection~master~I333c34063d47aff630c91879bcb4ad2fdae3c9df","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"10234":{"account":{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},"last_update":"2022-02-09 00:26:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I333c34063d47aff630c91879bcb4ad2fdae3c9df","subject":"Add Andrey Pavlov candidacy for Ec2 API","status":"MERGED","created":"2022-02-08 11:07:06.000000000","updated":"2022-02-09 00:28:45.000000000","submitted":"2022-02-09 00:26:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828279","meta_rev_id":"92a24b759d854260e15c615e4518887e963989a9","_number":828279,"virtual_id_number":828279,"owner":{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 00:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2022-02-09 00:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2022-02-09 00:15:50.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2022-02-09 00:15:50.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 11:16:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-08 13:52:54.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2022-02-09 00:00:56.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2022-02-09 00:15:50.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"146ed32676f6b2cb4cda75d992fb8a302fde94d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},"date":"2022-02-08 11:07:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"909b306f58c9c5656bbafdc6b628d979cb0c2e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 11:16:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf28afcbba349be9be6b6fc352aa2dd : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa30c5922963430e9ec045902612fb13 : SUCCESS in 3m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/99582001c35849d1986259f312107373 : SUCCESS in 3m 00s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/7431d5fb599c4950a4d7a82d600d2fa2 : SUCCESS in 4m 11s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e56858b3bf27496895ba8e49e0546c8b : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9d6de69c3a5a6cc906fc23a6e1bc81c2d00e746f","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-02-08 13:52:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9d4e19b4ee4e8c6efc92885c02eb7f38237af8f","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-02-08 23:07:29.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"beeb747277022d5a4cc5adb183e23e0737ddbfb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 23:18:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e9d2d5250b04c7ea1e4da53b890fcdf : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/074b61825a4449dbb1b1276f75a20532 : SUCCESS in 6m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d3234747e26840e5800b7f151c7d21df : SUCCESS in 3m 32s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/d2f31c08d61743598cd4bc16998d4fb0 : SUCCESS in 4m 00s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7573c67ac93649519cc4badb02dcd9d2 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"20cb916b2666ea772dd229dd46583abdd4f04376","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-02-08 23:56:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ee28ee899dcf741c4f36e0f53c5f9a72a59407b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-02-09 00:00:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3ebe39d962c28b86371195eb81786f1b81a1a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 00:05:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f460e9032fec44cb8d1219f44b34ae6d : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90f7377ba16a47d799138e51a1cda9a6 : SUCCESS in 3m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/040567f76304402ea3af45f26dffff49 : SUCCESS in 3m 49s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/ca6fd87c9a7e4889bd94978f4889e3c9 : SUCCESS in 3m 21s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/cc3099f954ca43d9a1ba1e35d284fbd9 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9b40c6eab92d9966e9f45c8fbff9f0e09c9d74de","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-02-09 00:14:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6bed9cd03d35bd274cf0a0bf6e51ac7cb9f58a4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-02-09 00:15:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9c244d67957c6909955bd115c22a673d634a6379","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 00:16:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82e7ab6db9970baf5e6d955295de0b9a1d13c715","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 00:26:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e2df7fc35642dca5a45334117bda59 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/523959dfe86c47de839366b8ce51588c : SUCCESS in 4m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/403c240273074320a23ab499d3f34b56 : SUCCESS in 4m 21s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a0558ef4e6c54dd78ca935e368cf89be : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"124498fbdcbdcfc5359ce72cc39bc0ae19867ae3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 00:26:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"92a24b759d854260e15c615e4518887e963989a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 00:28:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/8bfd1e155e93493687b3c91ae00b94f2 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8338deacdb43279ac904cfdca4264fa34868c749","revisions":{"8338deacdb43279ac904cfdca4264fa34868c749":{"kind":"REWORK","_number":1,"created":"2022-02-08 11:07:06.000000000","uploader":{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},"ref":"refs/changes/79/828279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/79/828279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/79/828279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/79/828279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/79/828279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/79/828279/1"}}},"commit":{"parents":[{"commit":"3b6c63deb94567aed7acd3ba1418c72b7614cbe0","subject":"Merge \"Set up candidacy directories for Z elections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3b6c63deb94567aed7acd3ba1418c72b7614cbe0"}]}],"author":{"name":"Andrey Pavlov","email":"andrey-mp@yandex.ru","date":"2022-02-08 11:06:47.000000000","tz":180},"committer":{"name":"Andrey Pavlov","email":"andrey-mp@yandex.ru","date":"2022-02-08 11:06:47.000000000","tz":180},"subject":"Add Andrey Pavlov candidacy for Ec2 API","message":"Add Andrey Pavlov candidacy for Ec2 API\n\nChange-Id: I333c34063d47aff630c91879bcb4ad2fdae3c9df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8338deacdb43279ac904cfdca4264fa34868c749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8338deacdb43279ac904cfdca4264fa34868c749"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
