)]}'
{"id":"openstack%2Felection~828466","triplet_id":"openstack%2Felection~master~Iff10f99fd21f3ccdbffeabc486d347dfd86f3c04","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"8543":{"account":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"last_update":"2022-02-09 16:21:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iff10f99fd21f3ccdbffeabc486d347dfd86f3c04","subject":"Add Lucian Petrut candidacy for Winstackers","status":"MERGED","created":"2022-02-09 07:41:19.000000000","updated":"2022-02-09 16:26:42.000000000","submitted":"2022-02-09 16:21:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828466","meta_rev_id":"e9cb7a0daa5079f17defc67d9a2ea22b97560aca","_number":828466,"virtual_id_number":828466,"owner":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 16:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2022-02-09 15:22:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-09 16:09:10.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-09 16:09:10.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 07:52:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 15:22:44.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2022-02-09 16:09:10.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"ee8aa3fc9811f01639d60779e4a5bf65213fd0cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-02-09 07:41:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"258cebdc6c270321ee2091e83caf07b26a4a220d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 07:52:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/390e4987e0514ef6b40ca7771876e005 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b239e974ab02493bb236a6d2f0ec0474 : SUCCESS in 4m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cc9eb77911454d04bda819f3728d27e8 : SUCCESS in 3m 57s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/a113334da5664c4e8719be697c18b708 : SUCCESS in 4m 02s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/456244cd2ff541cc811964eff4570c90 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"999b7aba259777c9b80a58b64d756903fd3fd063","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-02-09 15:22:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6db03392074a958e923b8fdcd8a5c4cb8d2503e3","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-02-09 16:09:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68086086e0ec7d12c133e3c68a0dc9b5b748d49a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:10:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84e456973bd4515d82157a73093040d204bed77d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:21:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a0408d396cf4268a5275843714aa591 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec9b3596e300468dad4b1b4c2f8fb0d4 : SUCCESS in 6m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2d40eb2834404fd4afada1d61729b531 : SUCCESS in 6m 38s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2da9a9df9d9f48449624cef4b0f698f2 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b34571fe12442a67de6c68dcb16b74819c1692e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:21:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e9cb7a0daa5079f17defc67d9a2ea22b97560aca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:26:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/d9d49e090c3347268373b5e25b44d528 : SUCCESS in 2m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ab3338cfa28c48c82a2b251782f6f499ff210ed","revisions":{"6ab3338cfa28c48c82a2b251782f6f499ff210ed":{"kind":"REWORK","_number":1,"created":"2022-02-09 07:41:19.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/66/828466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/66/828466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/66/828466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/66/828466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/66/828466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/66/828466/1"}}},"commit":{"parents":[{"commit":"b5de90e40dca497653372f70670a93a88f9a5234","subject":"Merge \"TC Nomination - spotz\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b5de90e40dca497653372f70670a93a88f9a5234"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2022-02-09 07:40:05.000000000","tz":120},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2022-02-09 07:40:05.000000000","tz":120},"subject":"Add Lucian Petrut candidacy for Winstackers","message":"Add Lucian Petrut candidacy for Winstackers\n\nChange-Id: Iff10f99fd21f3ccdbffeabc486d347dfd86f3c04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6ab3338cfa28c48c82a2b251782f6f499ff210ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6ab3338cfa28c48c82a2b251782f6f499ff210ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
