)]}'
{"id":"openstack%2Felection~829128","triplet_id":"openstack%2Felection~master~I8aabc67c8fa5df653492b0c2d8d4ccdefb633a5e","project":"openstack/election","branch":"master","topic":"tc_z","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2022-02-14 21:58:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8aabc67c8fa5df653492b0c2d8d4ccdefb633a5e","subject":"Add TC candidacy for Kristi Nikolla for Z","status":"MERGED","created":"2022-02-14 20:07:02.000000000","updated":"2022-02-14 22:00:48.000000000","submitted":"2022-02-14 21:58:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829128-tc_z","meta_rev_id":"703228e793722617f548fbbe83a561ceec50484b","_number":829128,"virtual_id_number":829128,"owner":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 21:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2022-02-14 21:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2022-02-14 20:28:57.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2022-02-14 21:51:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-14 20:12:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-14 20:28:57.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-02-14 21:51:14.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"c8db218a0172facb62c503c5e4129f916ddf9edc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-02-14 20:07:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dda4baa277b31a6a65091382d255919d9bc6bab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 20:12:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f32878664ea74526a373871a00182029 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08194d2a7dd443b49fd5601a97cf18b1 : SUCCESS in 3m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/944e831bc269468b91bbaee30ef19e38 : SUCCESS in 3m 23s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/064ac75907ec47e9a6ec04370aee2bdb : SUCCESS in 3m 15s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/bc48748a92d543d0822de77da7fc299f : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a5f07a29a70012e284b48a8559e8fd313d46a3cb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-02-14 20:28:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"310109e08aaf919fc90be384629b44d58576dcf2","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-02-14 21:51:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00927845fa95ffaca21f55841969911b80aaf4b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 21:51:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e626a35aa09a7085a3d827d3febcadd024cca3f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 21:58:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2239bd5fb1c45c0a2db2bdd658319b7 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa11cd7ebaf948c1ae7cb5a816bd5159 : SUCCESS in 4m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a567a71672c544a98d248fce22c4bd1e : SUCCESS in 3m 26s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a3881bddb2a344078d72271501697a70 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7bb2a4e9d284eb0f832baa91f9da4d1ba330da45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 21:58:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"703228e793722617f548fbbe83a561ceec50484b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 22:00:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/2e60ee89e4fa43da8e88ff9b981b7f0e : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea5417cf6b5bf41156d93395676da734c3696b5b","revisions":{"ea5417cf6b5bf41156d93395676da734c3696b5b":{"kind":"REWORK","_number":1,"created":"2022-02-14 20:07:02.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/28/829128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/28/829128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/28/829128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/28/829128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/28/829128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/28/829128/1"}}},"commit":{"parents":[{"commit":"7be931d052d9e904088d6167ead764589c583f18","subject":"Merge \"Add Pierre Riteau\u0027s candidacy for Blazar PTL (Z)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7be931d052d9e904088d6167ead764589c583f18"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2022-02-14 20:05:55.000000000","tz":-300},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2022-02-14 20:06:56.000000000","tz":-300},"subject":"Add TC candidacy for Kristi Nikolla for Z","message":"Add TC candidacy for Kristi Nikolla for Z\n\nChange-Id: I8aabc67c8fa5df653492b0c2d8d4ccdefb633a5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ea5417cf6b5bf41156d93395676da734c3696b5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ea5417cf6b5bf41156d93395676da734c3696b5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
