)]}'
{"id":"openstack%2Felection~830928","triplet_id":"openstack%2Felection~master~Id6b1011b26ab992ab988992c44e8e93dba88fc83","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-02-25 17:13:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id6b1011b26ab992ab988992c44e8e93dba88fc83","subject":"Update configuration for actual election end date","status":"MERGED","created":"2022-02-25 04:05:59.000000000","updated":"2022-02-25 17:15:47.000000000","submitted":"2022-02-25 17:13:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830928","meta_rev_id":"23febffe01044e1d8af033f33901fedfa2ea6078","_number":830928,"virtual_id_number":830928,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-25 17:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-25 16:18:08.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-25 17:02:08.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 04:16:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-25 16:18:08.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"71ca4f398ea712543fa447a45b9b7d161b91c7fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-02-25 04:05:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82abe69a910f5d9a58ec34f582e3e975718d21ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 04:16:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88044206c3af4e6fbd4da6e5500ce11b : FAILURE in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4295c263b75047a6ba74420f23888e20 : FAILURE in 3m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/610592269e1d49dcaace995312683f5f : SUCCESS in 3m 11s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/6b9d126baa3c48df9ab0fdc8e7e4ac1f : FAILURE in 3m 24s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/535b9a6061c948c98d927e111f8d8799 : FAILURE in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c5b79ca48a9f94c58899d9797675227fa327a207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-02-25 05:45:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d02d08c6e99b14f9b925edb91078cbda374bf52b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 05:52:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5601af842cc24937b320b9b7722ecef8 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e090f855d4ab42059b96b917390e51d3 : SUCCESS in 3m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9c090c11019e4b5e87c994c57a0a8ad1 : SUCCESS in 5m 20s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/af3f8157d76347c9a8e957f5e1d7fa63 : SUCCESS in 3m 10s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e339fe8edfca4dfa96774524f837aa7a : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"714fb14e5e905f0be5fd08e4cf0a611c21b85424","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-02-25 16:18:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b3361804eb43b6fc2f43519fc827bf0a11e0d37d","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-02-25 17:02:08.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e070bf50f8976540ed7d36bbee1084394be5ee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 17:02:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"917ea7485ae3a9887249381321c9db32d2f3fc0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 17:13:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50462e9bc29546778b07f02cd738a1ac : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a333166c0ff4d66a5d7b6bba72c6a7f : SUCCESS in 5m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/96b3557812994a90a6e856f15ed1902b : SUCCESS in 6m 11s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/02a27d9dd2f44393bd2c9ceb66dd066b : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":2},{"id":"75a98c23644a71d3020e37b2f9d5bc1c27e21f8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 17:13:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"23febffe01044e1d8af033f33901fedfa2ea6078","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 17:15:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/ed8d5f97fc614645b9e0e1861ead9d0e : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e7169ae86636038e2a101542971d2ab3fe0c4968","revisions":{"534a843105cbfefa118d3d2ad3cd5cb2d02cbafa":{"kind":"REWORK","_number":1,"created":"2022-02-25 04:05:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/28/830928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/28/830928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/28/830928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/28/830928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/28/830928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/28/830928/1"}}},"commit":{"parents":[{"commit":"44a8434f3bc67093b86e986c65e1d41ab4b35530","subject":"Close out Zed elections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/44a8434f3bc67093b86e986c65e1d41ab4b35530"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-02-25 04:02:57.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-02-25 04:05:35.000000000","tz":-360},"subject":"Update configuration for actual election end date","message":"Update configuration for actual election end date\n\nFor Zed election, no voting was required so we closed\nthe election after TC Campaigning period. Update\nconfiguration with actual election end date so that\nelection site shows the reality instead of election\nin progress which is confusing.\n\n-https: //review.opendev.org/c/openstack/election/+/830544\nChange-Id: Id6b1011b26ab992ab988992c44e8e93dba88fc83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/534a843105cbfefa118d3d2ad3cd5cb2d02cbafa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/534a843105cbfefa118d3d2ad3cd5cb2d02cbafa"}]},"branch":"refs/heads/master"},"e7169ae86636038e2a101542971d2ab3fe0c4968":{"kind":"REWORK","_number":2,"created":"2022-02-25 05:45:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/28/830928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/28/830928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/28/830928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/28/830928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/28/830928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/28/830928/2"}}},"commit":{"parents":[{"commit":"44a8434f3bc67093b86e986c65e1d41ab4b35530","subject":"Close out Zed elections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/44a8434f3bc67093b86e986c65e1d41ab4b35530"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-02-25 04:02:57.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-02-25 05:45:36.000000000","tz":0},"subject":"Update configuration for actual election end date","message":"Update configuration for actual election end date\n\nFor Zed election, no voting was required so we closed\nthe election after TC Campaigning period. Update\nconfiguration with actual election end date so that\nelection site shows the reality instead of election\nin progress which is confusing.\n\n-https: //review.opendev.org/c/openstack/election/+/830544\nChange-Id: Id6b1011b26ab992ab988992c44e8e93dba88fc83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e7169ae86636038e2a101542971d2ab3fe0c4968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e7169ae86636038e2a101542971d2ab3fe0c4968"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
