)]}'
{"id":"openstack%2Felection~852630","triplet_id":"openstack%2Felection~master~I650f9ef7933163c5de89d61e177111359b2679c1","project":"openstack/election","branch":"master","topic":"bauzas_nova_aa_ptl","attention_set":{},"removed_from_attention_set":{"15993":{"account":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"last_update":"2022-08-31 12:49:06.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-08-31 12:49:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I650f9ef7933163c5de89d61e177111359b2679c1","subject":"Sylvain Bauza candidacy for Antelope Nova PTL","status":"MERGED","created":"2022-08-09 16:06:17.000000000","updated":"2022-08-31 12:49:06.000000000","submitted":"2022-08-31 12:49:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852630-bauzas_nova_aa_ptl","meta_rev_id":"3c794e91dd4c441a747bbc8d9d68cf34fee53b6d","_number":852630,"virtual_id_number":852630,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 12:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2022-08-31 12:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2022-08-30 19:51:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2022-08-31 12:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-09 16:14:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 19:51:07.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2022-08-31 12:39:21.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"43c96182d13f2fded8f025177ec5bce570c36e3d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-08-09 16:06:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"904aadca95ee5e95951b48f0062603443a948a59","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-08-09 16:06:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"63aa538ec9ae5660d6b06fd1e3ff4225a1732725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 16:14:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7001803fce6849568b692cee18a3c81d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26053a9dbb24a6bb9c249705de68199 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5508c3f9cf9242eeb413260dad2f61b7 : SUCCESS in 2m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/77b4aefe9eb94aa2aad4b20521faaad7 : SUCCESS in 3m 33s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/d963ccc605814a50ac2a86bfb9c592ec : SUCCESS in 2m 55s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1fecf59ea1404a69990df591e2b00cd0 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"54dd59d87ab00d3b8e6649225c9e9d0641e8899a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-08-30 12:20:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"91fe1c03c4a0491eee201b700a29fd53950d4cde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 12:30:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb2942799b4e413caf1ca82827b58ab9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3253ecb8c7834fd68997fd988c970461 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f15d0460793a4835ad1f4b0bcd516e39 : SUCCESS in 4m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f955a3af6aa54185adf1f30db7c08f36 : SUCCESS in 3m 27s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/7337d77ee23c40928db93ccb6de66835 : SUCCESS in 3m 05s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/588a825d949e49e6bf82fadb37cda2e9 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"724938d997fbfbadcc1a5c2c789ae04324d681a2","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2022-08-30 19:51:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bd88e6d396fc4c13bffcdff05b8657f6caf86694","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-08-31 08:40:41.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"c7a357e0a4ba53a7b8783f712d9b76a0b29f2237","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-08-31 12:39:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b072d24fbff1e8bba5b2840eee8a36a5c3907427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 12:40:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f4375cca566e8e21a89db101ef888d2defd94f82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 12:49:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d23fa0a12c744aeb2310e6ff41d7f0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46c1136bfc4410d9c8b30db9d302914 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6738144f71f44fcc88653b394c503565 : SUCCESS in 4m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5966655071114421bdeba177adbf0dfc : SUCCESS in 3m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2a7a3e27db88471980c1233d253e5ef6 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"3c794e91dd4c441a747bbc8d9d68cf34fee53b6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 12:49:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c4495168948c2dd2d30365090a091af6c041de9b","revisions":{"7496da3cdd8d0112ebfd995bf23e0df7c1a03569":{"kind":"REWORK","_number":1,"created":"2022-08-09 16:06:17.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/30/852630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/30/852630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/30/852630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/30/852630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/30/852630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/30/852630/1"}}},"commit":{"parents":[{"commit":"6a06d4e6f054403d11449739efc9ec6e01bdcb30","subject":"Adding Sylvain Bauza candidacy for Nova Zzzzzz PTL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6a06d4e6f054403d11449739efc9ec6e01bdcb30"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2022-08-09 16:06:17.000000000","tz":0},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2022-08-09 16:06:17.000000000","tz":0},"subject":"[WIP] Sylvain Bauza candidacy for Antelope Nova PTL","message":"[WIP] Sylvain Bauza candidacy for Antelope Nova PTL\n\nChange-Id: I650f9ef7933163c5de89d61e177111359b2679c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7496da3cdd8d0112ebfd995bf23e0df7c1a03569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7496da3cdd8d0112ebfd995bf23e0df7c1a03569"}]},"branch":"refs/heads/master"},"c4495168948c2dd2d30365090a091af6c041de9b":{"kind":"REWORK","_number":2,"created":"2022-08-30 12:20:59.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/30/852630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/30/852630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/30/852630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/30/852630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/30/852630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/30/852630/2"}}},"commit":{"parents":[{"commit":"710f7b054b31109009a827833b76682a453412ac","subject":"Merge \"Add Eyal Bar-Ilan candidacy for Vitrage PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/710f7b054b31109009a827833b76682a453412ac"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2022-08-09 16:06:17.000000000","tz":0},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2022-08-30 12:20:47.000000000","tz":120},"subject":"Sylvain Bauza candidacy for Antelope Nova PTL","message":"Sylvain Bauza candidacy for Antelope Nova PTL\n\nChange-Id: I650f9ef7933163c5de89d61e177111359b2679c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c4495168948c2dd2d30365090a091af6c041de9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c4495168948c2dd2d30365090a091af6c041de9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[]}
