)]}'
{"id":"openstack%2Felection~854783","triplet_id":"openstack%2Felection~master~Ic19ab68bb9dd419cc7c7724413bb1698e06fe4aa","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"8064":{"account":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"last_update":"2022-08-27 23:49:55.000000000","reason":"Change was submitted"},"15993":{"account":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"last_update":"2022-08-27 23:38:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15993\u003e replied on the change","reason_account":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}}},"hashtags":[],"change_id":"Ic19ab68bb9dd419cc7c7724413bb1698e06fe4aa","subject":"Adding Jake Yip candidacy for Magnum","status":"MERGED","created":"2022-08-26 13:45:54.000000000","updated":"2022-08-27 23:51:33.000000000","submitted":"2022-08-27 23:49:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"854783","meta_rev_id":"efcf5a3c5a19743278642c7d7e53f07f36f19e53","_number":854783,"virtual_id_number":854783,"owner":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-27 23:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2022-08-27 23:36:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2022-08-27 23:38:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2022-08-27 23:38:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 13:54:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 20:18:12.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2022-08-27 23:36:12.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"962779df66cde7483de6d43467b1b12e796c0dfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2022-08-26 13:45:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81139d307a80e9b3180714f8abb74b91ae5290a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 13:54:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/288a432984444027b79f49eeed99e047\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f7f94db02624bdd9ad85996d42a926d : FAILURE in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abe5892924b045eb851cb72c350719bc : SUCCESS in 3m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00683cc548c649139444c03c1bfb9055 : SUCCESS in 2m 44s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/b35bc8dfb62b49a79491cc7a265f1cab : FAILURE in 3m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/593783de374b44fba847cb0148365fbd : FAILURE in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a6a084a867006d8f78c7a62d0a6f92ae22387f6d","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2022-08-26 14:29:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a53fd34962c602380229192f4f35d1da7e2f8059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 14:38:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70b23c6bf49b4d2fbe9a7bf423b6d7d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c09e84c5cfde4bbd8aa71352f8b0481a : FAILURE in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed6c5efd06064a1c9e602c1b07bc1c05 : SUCCESS in 4m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cc0c26e17f5c48769428edba257ef941 : SUCCESS in 5m 44s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/78e0dbb734c546da9e8b84f312a8d540 : FAILURE in 4m 18s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2dff184aed274f19b04cbd835bf197f3 : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9b0598941fe3aef54aed733cfdaf817ee6ccf78c","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2022-08-26 20:18:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"57cc738c6e792f2c68b25d1c6ee909dd510bb332","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2022-08-26 20:18:28.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_15993\u003e\n","accounts_in_message":[{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}],"_revision_number":1},{"id":"242f1370d33453a44755c472ffab96d7ea962219","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2022-08-26 20:19:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27fb001933df6488418c2f4d8357c607aded4f3d","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2022-08-27 01:33:55.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"86c85060a4690d09a46119765dc4b0a87f90e636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-27 01:43:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95109c74c3e847bbb1c8d3c0961a6282\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97df48ef0509422fb4a0355810d4b7a6 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7002a11d06f84a48a2a9cf4145e34038 : SUCCESS in 4m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/532141b87e484fa5a227f849526d2afa : SUCCESS in 3m 41s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/6f03b161a2324818afef626eeb8cf267 : FAILURE in 3m 33s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f0997206a3a446c5a1acc01d82153d60 : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0673c3226ac79aed2f7b0e5515d7316ca682242e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2022-08-27 02:13:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d736026d9edb226d804125152f3d59dcb00ee13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-27 02:37:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51637a89a28d475abada7ee8d7737fb0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037867ab5e8348a0935ff56114204d98 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d71fb271b664a7382683fa246eda3ab : SUCCESS in 4m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/967c7c5311af433ba942693ff68f7ddd : SUCCESS in 4m 29s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/54d0a8720f5d4806be4e525bc0cfdb60 : SUCCESS in 4m 32s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/8067d798dd6c47b9a784466db990e449 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d8adfc3f96005a1d0206621657a32bc55eb1cc1d","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-08-27 23:36:12.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ad8389c81322442148999177f56cdee661e5c04","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2022-08-27 23:38:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ca9b3389d91b206f154b7d441f25324d737f64d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-27 23:39:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7711aa89c937dd54738d416d3ac76ebd687e58b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-27 23:49:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf4c86d14de34a9c9e587e2b414b099e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380b994418614f13bbb7d6e0f2b1f46e : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06b1887fe6f145b0975038df53d300fa : SUCCESS in 3m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/30aa21f8f4bb4fe08cac349dd4e279c5 : SUCCESS in 3m 13s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/49c99154bb2942faa398ab33437cf74c : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4087e77f5f2d96b6687655e947c5e359b11f8dac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-27 23:49:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"efcf5a3c5a19743278642c7d7e53f07f36f19e53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-27 23:51:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b0a302de01407494c6000205ad4468\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/be5a0e658fd64781bcad8806948e0448 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7a5979d9bcc02b6195c9b2e92ca050b6a2a24eeb","revisions":{"8e8c34665faa4ad71c5f4e1358ea26aa77f88a49":{"kind":"REWORK","_number":1,"created":"2022-08-26 13:45:54.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/83/854783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/83/854783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/83/854783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/83/854783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/83/854783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/83/854783/1"}}},"commit":{"parents":[{"commit":"59d40211d45a75c5c8dff6d3530ffa3eed153804","subject":"Merge \"Start the Antelope Elections for PTL and TC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/59d40211d45a75c5c8dff6d3530ffa3eed153804"}]}],"author":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2022-08-26 13:43:47.000000000","tz":0},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2022-08-26 13:45:50.000000000","tz":0},"subject":"Adding Jake Yip candidacy for Magnum","message":"Adding Jake Yip candidacy for Magnum\n\nChange-Id: Ic19ab68bb9dd419cc7c7724413bb1698e06fe4aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8e8c34665faa4ad71c5f4e1358ea26aa77f88a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8e8c34665faa4ad71c5f4e1358ea26aa77f88a49"}]},"branch":"refs/heads/master"},"7a5979d9bcc02b6195c9b2e92ca050b6a2a24eeb":{"kind":"REWORK","_number":2,"created":"2022-08-27 02:13:17.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/83/854783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/83/854783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/83/854783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/83/854783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/83/854783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/83/854783/2"}}},"commit":{"parents":[{"commit":"a8502cf6f465ab80e35519a09e72985f2079ff67","subject":"Remove Ghanshyam from election official for 2023.1 election","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a8502cf6f465ab80e35519a09e72985f2079ff67"}]}],"author":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2022-08-26 13:43:47.000000000","tz":0},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2022-08-27 02:12:58.000000000","tz":0},"subject":"Adding Jake Yip candidacy for Magnum","message":"Adding Jake Yip candidacy for Magnum\n\nChange-Id: Ic19ab68bb9dd419cc7c7724413bb1698e06fe4aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7a5979d9bcc02b6195c9b2e92ca050b6a2a24eeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7a5979d9bcc02b6195c9b2e92ca050b6a2a24eeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}}]}],"submit_requirements":[]}
